)]}'
{"id":"openstack%2Faodh~808505","triplet_id":"openstack%2Faodh~stable%2Fvictoria~I6dac8447e50aabb97bdec65c61f46b9bc78002b7","project":"openstack/aodh","branch":"stable/victoria","topic":"cleanup_featuresets-stable/victoria","attention_set":{},"removed_from_attention_set":{"20182":{"account":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"last_update":"2024-03-01 08:59:59.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-13 21:59:40.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I6dac8447e50aabb97bdec65c61f46b9bc78002b7","subject":"Run TripleO jobs on CentOS8 instead of CentOS7","status":"ABANDONED","created":"2021-09-13 07:43:58.000000000","updated":"2024-03-01 08:59:59.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":808589,"cherry_pick_of_patch_set":1,"meta_rev_id":"6e737ee7a8187dae243e75f461e95301cd8acbb2","_number":808505,"virtual_id_number":808505,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-12-13 21:59:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"disliked":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"all":[{"value":1,"date":"2022-12-13 21:08:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":-1,"date":"2022-12-15 15:13:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 08:37:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 08:58:57.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-08-16 15:56:43.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"CC"},{"updated":"2022-12-15 15:13:47.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"}],"messages":[{"id":"9341940125c915df520cf730ac82159da2267ab0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-13 07:43:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"456603f0bf72a63f22ac6eee69630e02e39d943a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:37:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e26cf5b0434748a4afdfaf24a22c62 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5af6cee818d64a0cac6b9fc136313b1e : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40090062963848d5ae46670ba722f599 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0984ae78880040618cb79f6cb26d2af3 : SUCCESS in 8m 48s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/25d2eca73b294893b8e480fc9b82a22a : SUCCESS in 42m 00s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d3036673498f4b4795b1cf3c8e75634e : SUCCESS in 38m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e13640a8559946e9be27f3cee217b863 : SUCCESS in 51m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/81d313654d8a46ed9b99997ff2a4bc89 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4ed1a834cb54e567eb098967df013ec69545a20d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-14 12:33:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"389ed0b346140cdfa2133a4f7fdb9c60109b957f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 13:31:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f3005e5e004fa5b2337af1c9e7d361 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbdff68ba29a4864b2f7045b26f2b843 : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98a7fb98ddbe492b9866ca94431124f1 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92dff16c864d410a847454b4b330bff3 : SUCCESS in 7m 13s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9b88b4ccd3524a2ba74c0101193604e3 : SUCCESS in 38m 08s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d57d2d6152454c51a6caf9d072e7542f : SUCCESS in 40m 39s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d85d2e9a1b384672aae907bc53fb9adf : SUCCESS in 50m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2abe00051c7c447696d98049b59a3f18 : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a21421d78e0293a8e360f1be08dc0ac26f684933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 13:46:51.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ee3f0dc18e0dca029915f898f2a0d807718ba711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:54:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b1f75b996d42818eb9dfd3c88a45f2 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e8ca9083a8643a393331567272f296b : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/022a4117ca024b70b618e5d54314f528 : FAILURE in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af1fe40362f4f78bb934415a686712a : SUCCESS in 9m 29s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8c799ef72747402c9031e1674a357934 : SUCCESS in 44m 00s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/563b4b4205ac46b49acebeb5788524f3 : SUCCESS in 45m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/5ad0b2dc4690446a8542b65c769896fa : SUCCESS in 49m 54s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b50e00ab1bbb417c8f7839fc3030224e : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":3},{"id":"618f9941998c4f0013467bb00997a34888a4ac35","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:58:57.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65a925230c19cc4176059e6daf1823cbf539c235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 09:45:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f961e116f64a59979c5b866916d065 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/784a84bbd78149f2a3f1a68d58dc0795 : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fa1e2a2b1914ef4b32bacbcbeccb22b : TIMED_OUT in 41m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e99482802074138a4af7f6a5ff62488 : SUCCESS in 8m 00s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9e729fc0974d4e0480630d6e8301ac69 : SUCCESS in 34m 50s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/73b42e18ce2348b9837bc2c7cd5d4eaf : SUCCESS in 36m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b095a8ec57bc4cb1a1709e66082daa5e : SUCCESS in 32m 30s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f9176a1411d041879ef0f89966a4eb6e : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ed602560591eacbf201b48f4acc9df3e944620b6","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-16 15:56:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8dbcd53470d49766428604761794db81a4635385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 17:16:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d9b715c57a6498aac162547395be1b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f628a52a2aec41be8be93e9ad6c765b6 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d68ec192e044726a655eadbc490ddad : SUCCESS in 9m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/396bb7d591734ea1997955ceb9046f2a : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043ed893f00e4008a64cd741cb73f546 : FAILURE in 6m 06s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/eab34b2fd30b4fa7a3283732894e0d59 : SUCCESS in 44m 16s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/75f2cbd09b45472db5171de54231c7e3 : FAILURE in 52m 40s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/66dac137161e419cb81b3ca8d97381d9 : FAILURE in 45m 25s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a0b7e1396fc8425d92e370d7838347c2 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":3},{"id":"2a8712d9cb8e954e405301369342b78213daed0b","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-12-13 19:47:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dfe4fdccc5f7753ffdc269b46cd4ae8f3ff013a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 20:51:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23e8a23fdc584c6aba24838e29bbe40d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e36969ce2b043e887af41356a9c53a5 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cc47afe5cd04383b8768a0f51dddeec : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/806f5f48e96e46789cbcf52b16efbf27 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf05f558849404aaa9d6822f5b505d5 : FAILURE in 6m 31s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2167a47e490146d2bbd7941d293913ed : SUCCESS in 39m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/6b9938f750384c5b9ec77d3b3bba13d7 : SUCCESS in 44m 36s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/5c4d0154e46846dbbb55eab7456434bb : SUCCESS in 56m 24s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/451852317d2c44c89c74fa433a41f9fa : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":3},{"id":"0bcafbb26e2624e85198e8cfb9df7ae349d7d0fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-12-13 21:08:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"39c0d1e025971140b85fa1036ccf891d3d567895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 21:59:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b1d633aa31647c99092bdd317623078\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548addf6a3434e5c97c9a2a3befd9450 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8905e9df41342c4934793750221a1e1 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35c945a9b2314c8c9e7c7cba7b4031c8 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db7112cb34049248df724245bd67be4 : SUCCESS in 8m 01s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f45fae34b6c94339afd7fca10a12503b : SUCCESS in 39m 14s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/29cbc1d8082c4704b4b940e3cecda5df : SUCCESS in 41m 05s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bfad04c1ad08410197fee540261ad956 : SUCCESS in 45m 26s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0eff9ab0cd5b44c3be5afbb08bfd870c : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":4},{"id":"1bd7b77df67fc56cf488138fd61af9142538ae05","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-12-15 15:13:47.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6e737ee7a8187dae243e75f461e95301cd8acbb2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-03-01 08:59:59.000000000","message":"Abandoned\n\nAbandoned as part of transition from stable to unmaintained. Please repropose to unmaintained branch after creation if still valid interest.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"05a93e7ac52661fe17cd58140872c85909d5c103","revisions":{"42bbfd8651cb3b7a84be26ebfb62e4a261af8adb":{"kind":"REWORK","_number":1,"created":"2021-09-13 07:43:58.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/05/808505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/808505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/808505/1"}}},"commit":{"parents":[{"commit":"028561741e125f1b343ae5bf1021df1012e92005","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/028561741e125f1b343ae5bf1021df1012e92005"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:43:58.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/42bbfd8651cb3b7a84be26ebfb62e4a261af8adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/42bbfd8651cb3b7a84be26ebfb62e4a261af8adb"}]},"branch":"refs/heads/stable/victoria"},"7273abb8e9cf22d24253e1a0e7ecb2ce24c98050":{"kind":"REWORK","_number":2,"created":"2021-09-14 12:33:53.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/05/808505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/808505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/808505/2"}}},"commit":{"parents":[{"commit":"028561741e125f1b343ae5bf1021df1012e92005","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/028561741e125f1b343ae5bf1021df1012e92005"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-14 12:33:47.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7273abb8e9cf22d24253e1a0e7ecb2ce24c98050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7273abb8e9cf22d24253e1a0e7ecb2ce24c98050"}]},"branch":"refs/heads/stable/victoria"},"93aa32f1c807b6cd30871245671df63c96a40899":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-12 13:46:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/808505/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/808505/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/808505/3"}}},"commit":{"parents":[{"commit":"028561741e125f1b343ae5bf1021df1012e92005","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/028561741e125f1b343ae5bf1021df1012e92005"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 13:46:41.000000000","tz":540},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nBackport note:\nThe stable/xena branch is skipped because TripleO has no stable/xena\nrelease. TripleO jobs will be removed from stable/xena by a separate\nchange.\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n(cherry picked from commit d72944233f3d6abe7d68dc39710f4ccc0f7ac202)\n(cherry picked from commit 4b1b7855cb1d7ed453b1023c1842660699a8b5e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93aa32f1c807b6cd30871245671df63c96a40899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93aa32f1c807b6cd30871245671df63c96a40899"}]},"branch":"refs/heads/stable/victoria"},"05a93e7ac52661fe17cd58140872c85909d5c103":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-12-13 21:08:19.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/05/808505/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/808505/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/808505/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/808505/4"}}},"commit":{"parents":[{"commit":"7e73e6111e7173604bcb1ef28a285e4d47fbc72f","subject":"Replace oslo_utils.fnmatch with fnmatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7e73e6111e7173604bcb1ef28a285e4d47fbc72f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-12-13 21:08:19.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nBackport note:\nThe stable/xena branch is skipped because TripleO has no stable/xena\nrelease. TripleO jobs will be removed from stable/xena by a separate\nchange.\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n(cherry picked from commit d72944233f3d6abe7d68dc39710f4ccc0f7ac202)\n(cherry picked from commit 4b1b7855cb1d7ed453b1023c1842660699a8b5e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/05a93e7ac52661fe17cd58140872c85909d5c103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/05a93e7ac52661fe17cd58140872c85909d5c103"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
