)]}'
{"id":"openstack%2Faodh~808506","triplet_id":"openstack%2Faodh~stable%2Fussuri~I6dac8447e50aabb97bdec65c61f46b9bc78002b7","project":"openstack/aodh","branch":"stable/ussuri","topic":"cleanup_featuresets-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"20182":{"account":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"last_update":"2024-02-20 13:53:51.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-20 13:53:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6dac8447e50aabb97bdec65c61f46b9bc78002b7","subject":"Run TripleO jobs on CentOS8 instead of CentOS7","status":"ABANDONED","created":"2021-09-13 07:44:12.000000000","updated":"2024-02-20 13:53:51.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":808505,"cherry_pick_of_patch_set":1,"meta_rev_id":"690b24168e081972bdb3fc125986e9af80da1422","_number":808506,"virtual_id_number":808506,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-12 16:04:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"disliked":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"all":[{"value":1,"date":"2022-04-21 08:58:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":-1,"date":"2022-12-15 15:14:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 08:37:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 08:58:35.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-08-16 15:56:07.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2024-02-20 13:53:51.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"582557582419af04a8985a898052c5bc3d0a36f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-13 07:44:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"b1a168143427c0f9dd1d9dd1ea7f79acb3ce4e7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:37:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5362324eee61497bb775ae7082bf7451 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec4ccc6be4cc4aefa052f932f10540fd : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1822f4cdec84eddadb1b14f12955ccb : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cf607d0cc404b97af0fc86bb49fcf6d : SUCCESS in 7m 49s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5b033903e2c346dd9683c48e27ba9e4c : SUCCESS in 43m 21s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/c664574473cc48d3a97bb4ea573abe44 : SUCCESS in 50m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3edc8f24347a4ec4811f6530facc801b : SUCCESS in 39m 15s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b8251dc24937442d813547b031f0339c : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bdd9ff2819f0b2dd27c6de369af8a0e946d4b2b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-14 12:33:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c17d89bfe2dda645bc24b25d94e1c29e9c7bd5dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 13:18:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88704433212448cfab061ad793d43408 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90bad979a8ef4552b596f4fb3e3a49c8 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/132670dc1a7a4e9180aa92151a47e8f6 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d630d2dd21214c5db36526d2d7652231 : SUCCESS in 7m 44s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/cc0da3eb090b431b878ab7f59733e979 : SUCCESS in 33m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4a437283c613491aaee29c0aa5596370 : SUCCESS in 35m 38s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6b657571d9ea43509e9e2e3c326b0103 : SUCCESS in 39m 26s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/24f50e1f5d1541de877789d614b6c8f1 : SUCCESS in 43m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ff65ebe6d07a903dc0869ac80416263b5806290f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 13:47:15.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"59651160ddf9f8aab625809507a0f5b8e2c2aa3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:04:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871d3c9785084ae1911c6d4ab0a66360 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac64f8c6d6de4cac9fb64d4e79d78db9 : SUCCESS in 10m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/756a9cfbaf4447aea7ae03fdb1cd83a6 : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01931cbb31de4465966539c96e64d817 : SUCCESS in 9m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6ea74c39de2644f48f969b3f5f0941ec : SUCCESS in 48m 29s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/33da5d2d00bf4154b2c8b4badcd90dac : SUCCESS in 59m 59s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2da32bf6abdb4fff90a68f26e8e5ea22 : SUCCESS in 48m 25s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/511ed8d9e43d471e91fc8d241b40b5ce : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c3738757e923cbc0b9ae208b740a97781e91125a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:58:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1deea6be08978a41332df7cc12af5a6e50b5e2de","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-16 15:56:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"60970a9cefc70da9c94e1efbe8428cb3e3177222","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-12-15 15:14:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fe35cc2968bf45e927c6473d8409dd80546dfc50","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-12-15 15:14:44.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"690b24168e081972bdb3fc125986e9af80da1422","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-20 13:53:51.000000000","message":"Abandoned\n\nstable/ussuri branch of openstack/aodh transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87500b06b4fc179fbc941ff83ef05d0279eb612b","revisions":{"4d3d394dbb1250940f57d7e6c377c5d479606599":{"kind":"REWORK","_number":1,"created":"2021-09-13 07:44:12.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/06/808506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/06/808506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/06/808506/1"}}},"commit":{"parents":[{"commit":"d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:44:12.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4d3d394dbb1250940f57d7e6c377c5d479606599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4d3d394dbb1250940f57d7e6c377c5d479606599"}]},"branch":"refs/heads/stable/ussuri"},"8e34fd3fa8afc10f391628c9f87d6f6890cdde8d":{"kind":"REWORK","_number":2,"created":"2021-09-14 12:33:21.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/06/808506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/06/808506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/06/808506/2"}}},"commit":{"parents":[{"commit":"d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-14 12:33:15.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8e34fd3fa8afc10f391628c9f87d6f6890cdde8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8e34fd3fa8afc10f391628c9f87d6f6890cdde8d"}]},"branch":"refs/heads/stable/ussuri"},"87500b06b4fc179fbc941ff83ef05d0279eb612b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-12 13:47:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/808506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/06/808506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/808506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/06/808506/3"}}},"commit":{"parents":[{"commit":"d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 13:47:05.000000000","tz":540},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nBackport note:\nThe stable/xena branch is skipped because TripleO has no stable/xena\nrelease. TripleO jobs will be removed from stable/xena by a separate\nchange.\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n(cherry picked from commit d72944233f3d6abe7d68dc39710f4ccc0f7ac202)\n(cherry picked from commit 4b1b7855cb1d7ed453b1023c1842660699a8b5e6)\n(cherry picked from commit 93aa32f1c807b6cd30871245671df63c96a40899)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/87500b06b4fc179fbc941ff83ef05d0279eb612b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/87500b06b4fc179fbc941ff83ef05d0279eb612b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
