)]}'
{"id":"openstack%2Faodh~808507","triplet_id":"openstack%2Faodh~stable%2Ftrain~I6dac8447e50aabb97bdec65c61f46b9bc78002b7","project":"openstack/aodh","branch":"stable/train","topic":"cleanup_featuresets-stable/victoria-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"20182":{"account":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"last_update":"2023-12-21 14:12:50.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-21 14:12:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6dac8447e50aabb97bdec65c61f46b9bc78002b7","subject":"Run TripleO jobs on CentOS8 instead of CentOS7","status":"ABANDONED","created":"2021-09-13 07:44:30.000000000","updated":"2023-12-21 14:12:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":808506,"cherry_pick_of_patch_set":1,"meta_rev_id":"b39790872eb90ea5e19dd4236c9fd38f28e54796","_number":808507,"virtual_id_number":808507,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-12 16:48:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"recommended":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"all":[{"value":1,"date":"2022-04-21 08:58:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-12-20 11:31:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2022-12-13 19:35:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-16 13:35:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 08:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 08:58:24.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-08-16 13:35:43.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"state":"REVIEWER"},{"updated":"2022-12-13 19:35:19.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2022-12-20 11:31:22.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"31295be8cb3da1e5d9e5478f8fb209c43103c52b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-13 07:44:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"d82356775c7ac51e547c2ec1a4ff9f2f78ab9550","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:48:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8070745c6e5a4669b0089689f4996256 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d21e9af3aad4fc698e8d3e956239ead : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9e3945c9f4c4c9e8ac2e04c6040a23e : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04cc461d1f5a41d5a797e5ebf73e72d7 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37988d1fe92c45a0ac264f8c0756bde0 : SUCCESS in 7m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b8952bb5f5804046967943ba1c55ad27 : SUCCESS in 39m 11s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/bb69dfd402ef4ad68bd26315678065c3 : SUCCESS in 43m 11s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/fb4fe54b71e043ad8bf68c011585b94c : FAILURE in 49m 28s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1d31780845714db78f2987deb09d3297 : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ecc7601190ec7bc67d23b59203a1a5fcc8bbbb64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-14 12:32:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0b73ee063b983a91b9dcdb4c7c72529d7da297e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 13:38:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78136ba9a8d94772963b8cf4eb1c2f98 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18583768c4e241d481a68e9072403d23 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44d87496823d4fe4921abf2b67714f78 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a588d876a9e478db22614e23551e930 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9189e0664d1f4631a6c8c0ce01551aed : SUCCESS in 7m 38s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e504ec0ed16d4c9dadcc5684030cae4d : SUCCESS in 33m 22s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/93347c64edd64c68b46af078fa6d621d : SUCCESS in 42m 45s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f213016d55b94356935b92eabf060334 : SUCCESS in 1h 00m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/596da31eae8a4ce6bb1cbeb3475daf64 : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b055cc86443c62b181e5a2fa7e30d8e7b9f78a5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 13:47:49.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"199a0e3e80256ad720882786a5e32b1eea8fa7f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:48:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f6e5b7b99324e3196ce58aae506530a : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c84e6f95f71f47f59aa724e1ab19439e : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbaae45b0f3e455d887ce613edfb2ed4 : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7ef272db94641df8b02c2129600547d : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c4872fd4a6140e49300555183b612e4 : SUCCESS in 8m 15s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/47b66cc4d1634bd28d601a217f754863 : SUCCESS in 48m 42s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1fbb2a75f3b64cb39457875573869b39 : SUCCESS in 49m 35s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/620d452864f2453d896b690a07560c0b : SUCCESS in 1h 07m 33s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/83787e2033fa4b21a2fbc24df19f944d : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8a25f5498c332014a9fb5bd226334be60e76b502","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:58:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3d3ef4508c9e97a09df6269a913b39c93b49548f","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2022-08-16 13:35:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c597d8fb574f1e596ce85b3e143a94f9820b67be","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-12-13 19:35:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cf18e37bf5182c5a3520b30ae7376dae384f38ea","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-20 11:31:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b39790872eb90ea5e19dd4236c9fd38f28e54796","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-21 14:12:50.000000000","message":"Abandoned\n\nTrain is about to transition to End of Life. Open patches needs to be abandoned before branch deletion.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8ec5f39d1384024a9b99c7a3ed7b95a14c2e4306","revisions":{"472737eeb6204937696f7ac923e2ca1924e69043":{"kind":"REWORK","_number":1,"created":"2021-09-13 07:44:30.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/07/808507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/07/808507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/07/808507/1"}}},"commit":{"parents":[{"commit":"5415b96d1e58615fa80fee2eb74dc9cda90d8e2a","subject":"Use config options when creating keystone session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5415b96d1e58615fa80fee2eb74dc9cda90d8e2a"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:44:30.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/472737eeb6204937696f7ac923e2ca1924e69043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/472737eeb6204937696f7ac923e2ca1924e69043"}]},"branch":"refs/heads/stable/train"},"aef38b0cedb911a23944e6afd2c0f83c548f6044":{"kind":"REWORK","_number":2,"created":"2021-09-14 12:32:30.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/07/808507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/07/808507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/07/808507/2"}}},"commit":{"parents":[{"commit":"5415b96d1e58615fa80fee2eb74dc9cda90d8e2a","subject":"Use config options when creating keystone session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5415b96d1e58615fa80fee2eb74dc9cda90d8e2a"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-14 12:32:24.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aef38b0cedb911a23944e6afd2c0f83c548f6044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aef38b0cedb911a23944e6afd2c0f83c548f6044"}]},"branch":"refs/heads/stable/train"},"8ec5f39d1384024a9b99c7a3ed7b95a14c2e4306":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-12 13:47:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/808507/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/07/808507/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/07/808507/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/07/808507/3"}}},"commit":{"parents":[{"commit":"5415b96d1e58615fa80fee2eb74dc9cda90d8e2a","subject":"Use config options when creating keystone session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5415b96d1e58615fa80fee2eb74dc9cda90d8e2a"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 13:47:40.000000000","tz":540},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nBackport note:\nThe stable/xena branch is skipped because TripleO has no stable/xena\nrelease. TripleO jobs will be removed from stable/xena by a separate\nchange.\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n(cherry picked from commit d72944233f3d6abe7d68dc39710f4ccc0f7ac202)\n(cherry picked from commit 4b1b7855cb1d7ed453b1023c1842660699a8b5e6)\n(cherry picked from commit 93aa32f1c807b6cd30871245671df63c96a40899)\n(cherry picked from commit 87500b06b4fc179fbc941ff83ef05d0279eb612b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8ec5f39d1384024a9b99c7a3ed7b95a14c2e4306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8ec5f39d1384024a9b99c7a3ed7b95a14c2e4306"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
