)]}'
{"id":"openstack%2Faodh~808589","triplet_id":"openstack%2Faodh~master~I6dac8447e50aabb97bdec65c61f46b9bc78002b7","project":"openstack/aodh","branch":"master","topic":"cleanup_featuresets","hashtags":[],"change_id":"I6dac8447e50aabb97bdec65c61f46b9bc78002b7","subject":"Run TripleO jobs on CentOS8 instead of CentOS7","status":"MERGED","created":"2021-09-13 07:36:51.000000000","updated":"2021-09-29 08:39:31.000000000","submitted":"2021-09-29 08:32:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808589-cleanup_featuresets","meta_rev_id":"ddd9982031407469e41f159d9d335c4a11cf5bf1","_number":808589,"virtual_id_number":808589,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-29 08:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2021-09-20 05:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2021-09-28 05:50:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 08:29:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 09:58:19.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2021-09-20 05:57:23.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"995e324ab4d2ef43299e5167056bf7b2a9a6df63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-13 07:36:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edd0b856e3fd6e4b873f18fc02918d1225094e99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:29:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db1e3575293d4c10a1e586714c54486e : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/441703425313466aa1f3006ffc7aabfc : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dd27db16e8244369153124861bc357e : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a097df18156146f3a6e923cea5ccb5a0 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39c069dd4dd14662ad50479561b7fee5 : SUCCESS in 9m 56s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8d583c7e06f4471bacdd8fe069c189c2 : SUCCESS in 47m 56s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/220b19dfe50142bbaa996420b084ab9e : SUCCESS in 42m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/644f1fdce70b4c7eb21af0a218e7a67b : SUCCESS in 43m 40s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/67eb979983d54e67a77a62f2e8a41ef0 : SUCCESS in 47m 20s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/1badb008c043417a9780ca50204d669b : SUCCESS in 50m 19s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/75aa7cfd01a54341a20063132dea2205 : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":1},{"id":"00affc305ef89caf2b98229e4cfc59394275b2e6","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-09-13 09:58:19.000000000","message":"Patch Set 1:\n\nThank you for your patch.\n\nIt would probably be more honest and not misleading if you\u0027d just remove these. As you see, the checks are not being executed in CI at all, what do you think?","accounts_in_message":[],"_revision_number":1},{"id":"f180ea10215601242254b13355ed27afde93b02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-14 12:31:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f90b5f9e85f5ab1084f3e503f4d18c80b049cea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 13:31:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9549ee7a884eba8ce3881f2cf96ea5 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e241d5b49ab4661b7a28c7939dceba3 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea2584a1044847a7805b081a95db000c : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9f1c42e9ee24afcbdbcb40cab5ce4e9 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869b91c7aba741bcb1f1a4cd8687c05b : SUCCESS in 7m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e47a3d34fdf44ded8c3c817f8ac8fe6a : SUCCESS in 36m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/456a4852c7cd4e60af2f7db0be5a3350 : SUCCESS in 36m 20s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2dd9480e8c4545f791bed6dd9b36c9b4 : SUCCESS in 41m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b097f17e618e476f9f1403dfc7706934 : SUCCESS in 39m 12s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/3c147079ce104e2c8bfd9cc9c421c2b1 : SUCCESS in 42m 52s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/267558276255458186640aed988e5df4 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":2},{"id":"de5994f113c1df9521c5ba6cac64e5e282c68f77","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-09-16 07:09:11.000000000","message":"Patch Set 2:\n\nSAme comment as before applies. The tripleo jobs are not being executed at all. I didn\u0027t had the time to look at \u0027why\u0027.","accounts_in_message":[],"_revision_number":2},{"id":"32f2eaebf3004fe258b19618e44aa04918c19214","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 07:35:50.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Thank you for your patch.\n\u003e \n\u003e It would probably be more honest and not misleading if you\u0027d just remove these. As you see, the checks are not being executed in CI at all, what do you think?\n\nHi Matthias, Thanks for review!\n\nThe TripleO job executed only when we hit the \"experimental\" in the comment on patch specific patch that we want like\"recheck\", in short it works exactly the same way we do the \"recheck\" patches.","accounts_in_message":[],"_revision_number":2},{"id":"1ebd63f188138954fbb814b728b743ca7f281680","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 07:39:28.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e SAme comment as before applies. The tripleo jobs are not being executed at all. I didn\u0027t had the time to look at \u0027why\u0027.\n\nThe TripleO job executed only when we hit the \"experimental\" in the comment on specific patch that we want, like \"recheck\", in short it works exactly the same way we do the \"recheck\" on patches. I think it\u0027s safe to keep the TripleO job. Developer can test there change related to TripleO any time by hitting the \"experimental\" keyword on specific patch.","accounts_in_message":[],"_revision_number":2},{"id":"564c5d45116c99846a9bebf6dd48ce166cb69e2b","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 07:51:11.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Thank you for your patch.\n\u003e \n\u003e It would probably be more honest and not misleading if you\u0027d just remove these. As you see, the checks are not being executed in CI at all, what do you think?\n\nHere you will see the example:\n\n```\nMarios Andreou\n\ncheck experimental\n\n2 Jan 04, 2019 7:49 PM \nZuul\nBuild succeeded (experimental pipeline).\n\ntripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/02/620302/2/experimental/tripleo-ci-centos-7-scenario001-standalone/62b1c92/ : SUCCESS in 1h 14m 29s\n\n```\n\n[1]: https://zuul.opendev.org/t/openstack/builds?job_name\u003dtripleo-ci-centos-7-scenario001-standalone\u0026pipeline\u003dexperimental\n[2]: https://review.opendev.org/c/openstack/aodh/+/620302/2#message-24a2e1a72dde2755a3526841453f2b5a0141236b\n[3]: https://review.opendev.org/c/openstack/aodh/+/620302/2#message-75b974dc3b9960bea234fcb8c7e620ce6153854d","accounts_in_message":[],"_revision_number":2},{"id":"23e184f5fdb3dcaf5e59aeb96389673720f11af8","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 08:25:08.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"325fbdc6de2e05f0aa50fa7e9445c842b1a50e93","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:49:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a61c6420a40a466792781007c7f4b68a : SUCCESS in 2h 12m 20s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/6bf91fa5222b41aa9ff0f5c6131146b6 : SUCCESS in 1h 46m 12s","accounts_in_message":[],"_revision_number":2},{"id":"baebc1d3f35d2308a31630153ce0ac9fefe88c3d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-09-20 05:57:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"12df40081eed25feb8c9b05eeeb36d103aa37b2d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-09-28 05:50:10.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d8741a08abcc5fc08e3cfb72dda49e3814786892","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 05:50:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3db812b333bee7766eba5fc9f2cf9e5a6533f93e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 07:02:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e973e1acefd8409f89a75fa41229e599 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61ddd03d6361475ebe682ee691019257 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bb1df64fd40459f9d7e75e36b898357 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9853f463b61e4910abbd07e0812eb3b2 : SUCCESS in 7m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/50fc57331b624fe1b55a4ecbc55b1f74 : FAILURE in 37m 36s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a9ccdb585fcd4a4081479e9e822bb8c1 : FAILURE in 1h 04m 45s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6d53ca05c9214aaea20ea667dc57c082 : FAILURE in 41m 05s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f354796edb244c17a3354e962684f8c0 : FAILURE in 41m 19s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/f1c0a8f753454fa4b42dddf9f66554dc : SUCCESS in 58m 19s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/5662a51d7c7547e698207b2bb1a32c8b : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":2},{"id":"37eb625d59a674bf5f86258aa9ad40e570cee191","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-28 07:19:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9a329fd01d5ec58b7ba5566ab40e1fb6951f07b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 08:48:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2246e6b36ba4460280b618503ae77f61 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b43455cf1fe4eadb4562ca769de0f07 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5575e09274cc4f52958078a994a1ff76 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21b4440dbaa9450380900bb7859106fd : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ff99ab1c26429899c45a8dac60061a : SUCCESS in 6m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/87fee022c4b14a858cf15fe508db7b03 : FAILURE in 32m 57s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d930680ab0434b70ae6c16c10ee27ffd : FAILURE in 33m 40s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/54d1d4d343a84ecea02559141ee21b4e : FAILURE in 55m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a62294bcf64848db8dfe3ac59126046e : FAILURE in 30m 12s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/ee2640d7658344ec962ac3d2f1dc8615 : SUCCESS in 1h 05m 03s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/1b5a29509f074286b731b54bf7b8e204 : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":2},{"id":"841516f71992eb01d67e9732cfbe6a1eaf97e8ff","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-09-29 05:53:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c6b5f8c1a20347992d31b375fae1dfcdd77aa987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 07:06:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626c81dc4e1c4a1d9ed5b0ed54be2002 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b164ad9c6949908b6e8403121e3212 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7976add43e143b2b4fa8021a09772ca : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97f4e78b989f42718945cf47b1f980ba : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed75aec760f40caba40e733c604862d : SUCCESS in 7m 16s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0d5db2ca47124595b0e81f6d04782f50 : SUCCESS in 49m 42s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a30cd8ac662f431eb28d3111c7ef5818 : SUCCESS in 50m 02s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c3cf43bc83b2491f8bc0ccc43f693fcf : SUCCESS in 46m 58s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8d44197729cf4f1da56b1b071fc6969b : SUCCESS in 40m 42s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/f5a08b57a8da422b8bc34158d4dfe5ca : SUCCESS in 58m 27s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/aefbf699dfb04b98a8f8d6b4741b85d8 : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"8be476213a11531e3c8c7d5c39b08af243588639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 07:09:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a5e00e203bcb57f2e224fd6097dbffc850cfddf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 08:32:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe1690d6da9c44fb948f9f3c32b0bb2e : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ad9f313e72d434dbbb51b2226acd06d : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae9233f9e29e432da70584a32d9ed04f : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79bab14e0bde4020a2879d0fec9727a5 : SUCCESS in 11m 08s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9fad490137164dd5973a1d7aa95b0d6f : SUCCESS in 36m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b984de53234b424da33a133773ff3e1f : SUCCESS in 40m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/27c9b70027434dc8afbd2992a233a336 : SUCCESS in 49m 30s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/abeb5cc779884d28b94df2d67494c9bd : SUCCESS in 46m 57s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/a5117b8c520f4f9e90dbf13f37f5142f : SUCCESS in 52m 03s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/77d986515396460ab758bd636471392f : SUCCESS in 1h 15m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d15eb0fa6b05695c118cf5b05ed7891de600d344","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 08:32:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ddd9982031407469e41f159d9d335c4a11cf5bf1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 08:39:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4bd22f12c62485c95477f09978fb794 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d72944233f3d6abe7d68dc39710f4ccc0f7ac202","revisions":{"e8a1f27f96d1d0ad58ea2a7370720494030c7f86":{"kind":"REWORK","_number":1,"created":"2021-09-13 07:36:51.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/89/808589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/89/808589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/808589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/808589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/808589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/89/808589/1"}}},"commit":{"parents":[{"commit":"e61fba4a0bce1305366071addfba0b4cb290ad8f","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e61fba4a0bce1305366071addfba0b4cb290ad8f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:36:47.000000000","tz":330},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e8a1f27f96d1d0ad58ea2a7370720494030c7f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e8a1f27f96d1d0ad58ea2a7370720494030c7f86"}]},"branch":"refs/heads/master"},"d72944233f3d6abe7d68dc39710f4ccc0f7ac202":{"kind":"REWORK","_number":2,"created":"2021-09-14 12:31:37.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/89/808589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/89/808589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/808589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/808589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/808589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/89/808589/2"}}},"commit":{"parents":[{"commit":"e61fba4a0bce1305366071addfba0b4cb290ad8f","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e61fba4a0bce1305366071addfba0b4cb290ad8f"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-13 07:35:43.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-14 12:31:32.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I6dac8447e50aabb97bdec65c61f46b9bc78002b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d72944233f3d6abe7d68dc39710f4ccc0f7ac202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d72944233f3d6abe7d68dc39710f4ccc0f7ac202"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
