)]}'
{"id":"openstack%2Faodh~819898","triplet_id":"openstack%2Faodh~master~I4a6194fbf6a5317a523d75e3fb6bd2fd055f5100","project":"openstack/aodh","branch":"master","topic":"paste-ini-conf","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-06 02:32:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a6194fbf6a5317a523d75e3fb6bd2fd055f5100","subject":"Load api-paste.ini from configuration directories first","status":"MERGED","created":"2021-11-30 15:58:54.000000000","updated":"2022-01-06 02:36:03.000000000","submitted":"2022-01-06 02:32:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819898-paste-ini-conf","meta_rev_id":"de9637e642ce1bd9c63ebfec2a43fc4ba80ed59a","_number":819898,"virtual_id_number":819898,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-06 02:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2021-12-22 12:09:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2022-01-06 01:15:34.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2022-01-06 01:15:34.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 17:02:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-22 12:09:26.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2022-01-06 01:15:34.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"280e5a4718494dd09ac82dd90130c58cd0f1e63d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 15:58:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"113ea4a4cad0564d0ac875ccdd5bb563764d5920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:02:36.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3accc3f5bd4c5c8dc78bb0fce5bea1 : FAILURE in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a78464822e264fd1bd850f368afe2394 : FAILURE in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d5cbfbd2e464dacbcc3811dfe543ab4 : FAILURE in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e388f16d92464daadba7ad20c3041a : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47bd6f7fe8a646b28d9874f7f64178ea : SUCCESS in 8m 13s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ee6c675f4c204dc992b3538c2e6844e9 : FAILURE in 38m 44s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4c139b2df6f54b2a8ea1051a12dd4e61 : FAILURE in 39m 40s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/98bdee07e3e44ea0b4a65eabeba0713b : FAILURE in 40m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/83c0713f4e51449aa373194bc6abb7c2 : FAILURE in 40m 36s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/fb37f83613f34769bb82e257825e05c1 : FAILURE in 57m 35s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/4e77fcc39b16463782b52a41c1404434 : FAILURE in 51m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bfd276b3a284948cbcb89be2dac20b0e1deac163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 22:56:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aea0c47da1e6f07dfe01bec5408d486d8d466415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 23:41:56.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b913c18418471f9f287fc6bcc2c18d : FAILURE in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d143d688d09e46b28eda9ec80ccd48d0 : FAILURE in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e073160031f44e7a913d4be1b9425c1 : FAILURE in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9d53a0899444d5843fe65ae8c8a22e : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67a20cb5fd2942babac3596b0d4362be : SUCCESS in 4m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/157f6a8b49d34dcc9f2cac47384ca883 : FAILURE in 35m 51s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/51eb0fd672824dbc90692c0c79d4d7cf : FAILURE in 31m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/312a20da154f4084acd4a37259854374 : FAILURE in 30m 19s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/baee3cb40add41d8b6d32607d684f93e : FAILURE in 28m 28s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/e01ab4d773504d51b2ecc815630ba01a : FAILURE in 43m 46s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/f96a8483d43147c1ac149931bb930c01 : FAILURE in 34m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4d139d1397b586a32410ddfd803fe282a7580f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 23:45:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"85e2614450a1d266a44e316697a5c74fb070761c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 23:48:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5162c7d58ae16c396cdf98d16defd83d60eef5e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 00:40:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ac52e01ded4a7aadb5b37408a3738e : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61453bf97da840709fbbcb4f94b6e53b : FAILURE in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81f31e4ee53940d7aafcebc0e7330328 : FAILURE in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52063f7b956b4e7da1a83c3637357a39 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/559d6e29bf614eb2aca74087436c60ae : SUCCESS in 4m 48s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/022f3c9664aa4ec2aa902103ccf8b1ec : SUCCESS in 27m 21s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3a8ff30a8875404eb5b03e0b3b7bf1d2 : SUCCESS in 25m 28s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e24400d245a94b74899396f6ce35ee2f : SUCCESS in 28m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/7970db12758f4a0db189bcd8a7726565 : SUCCESS in 40m 50s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/b304a53b3d7f46ea82067cac690f2876 : SUCCESS in 37m 03s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/1187da152cc54e869e22bc7787de5256 : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":4},{"id":"ccf3fda69c1a5335476da178ca2d8dfcfb570a90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-01 01:35:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1f4ba307cbd7e2590d8da7a5a78530abf2762e45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-01 01:50:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9576627390256e3f1f62f32b10415725369c9ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 02:57:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ddec0e78c146feb0c3aab4410bfbd9 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/945979b5697b4ae9ab57afbaaf87c2ec : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e611ba60d49145fabac8a91473637b84 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f9803c6828417083803da15e83bba9 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1104747c6d5446a8af408e5d5e4eb80 : SUCCESS in 5m 38s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/55a9bc5e9e9044b69d4cec0250e30086 : SUCCESS in 42m 45s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/cd2a43c204f243b0a6e0d22388ee194c : SUCCESS in 45m 39s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b1807becddbb43af9a5a8635727af0b4 : SUCCESS in 50m 30s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e225f2e7b2a9473f86d45ee84200a89e : SUCCESS in 48m 32s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/31a6eede48354386b2c2ee88c49dd041 : SUCCESS in 59m 02s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/3ef0e7c73c8c4e1db84fba477fe89d3f : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":6},{"id":"54436b80443031ddaa0d8b346f298ee751c9c143","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-12-22 12:09:26.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"49aa46e63969f6d2aa7305c6515878deab5c7bbf","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-01-06 01:15:34.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"248c22455ab059ab0e3bfcfac50db704aa92753e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 01:16:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"eb1a4900553ab2fbaa9cafe6c8dd4b0cb614528f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 02:32:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701b507e5b3b4862a6760d15a2fe7c17 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f449c0251a74053b8722cde5059e4ff : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7fbe73e44e54edca5f27e9aea3a242e : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7487c7ab28bb4273b0f5ca05df8fcdbd : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba3aacc742a1412ca93994421e76fb1d : SUCCESS in 6m 20s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5bd1fe94a8ca49b3a3654e67c581e5d7 : SUCCESS in 42m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a40d4061336e4e80bcc2a785f416357c : SUCCESS in 48m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/55d5e0d405ac465e9aa64df9fca1ccf3 : SUCCESS in 39m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0568ec37b70a4d3ab096ada135b30363 : SUCCESS in 52m 18s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/a0105ca4b8f34f13afa8d2a70c3c925d : SUCCESS in 1h 00m 19s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/cea9b7b6bd4f4ff79d5739e11e09ab02 : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":6},{"id":"7d0783b6533fae8ce21b30bf6a10fbf2344319ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 02:32:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"de9637e642ce1bd9c63ebfec2a43fc4ba80ed59a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 02:36:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb2e814273e4085aac6380529a62fa0 : SUCCESS in 2m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5efd3f5113cd4c0bb1ccbb789776bd96 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8ac7c65cffac34e412ab117f8e86b4dd2c973ac9","revisions":{"18e196eafc6cc3aa82fff99cc725b24d4f682119":{"kind":"REWORK","_number":1,"created":"2021-11-30 15:58:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/819898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/819898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/819898/1"}}},"commit":{"parents":[{"commit":"6bbf61a67ba70726f63ff7e287e0e6ad30e4af06","subject":"Merge \"Rename [coordination] heartbeat to hearbeat_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6bbf61a67ba70726f63ff7e287e0e6ad30e4af06"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 15:43:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 15:58:35.000000000","tz":540},"subject":"Load api-paste.ini from configuration directories first","message":"Load api-paste.ini from configuration directories first\n\nThe api-paste.ini file defines WSGI definition of API. This file is\nregarded as one configuration file and sometimes users need to modify\nthe file, for example to enable an additional middleware.\n\nHowever currently aodh-api by default expects the file is located in\nthe aodh source directory and users should set the [api] paste_config\nparameter in addition to copy and modify the file.\n\nThis change makes the aodh-api process search the api-paste.ini file\nfrom configuration directories automatically, so that users can more\neasily customize the file by putting the modified file in /etc/aodh\n(or any other config directories).\n\nIf the paste_config parameter is defined as a full path then aodh-api\ndirectly loads the file from the specified path. If the file is not\nfound then the default file in the aodh source directory is used.\n\nChange-Id: I4a6194fbf6a5317a523d75e3fb6bd2fd055f5100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/18e196eafc6cc3aa82fff99cc725b24d4f682119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/18e196eafc6cc3aa82fff99cc725b24d4f682119"}]},"branch":"refs/heads/master"},"6f95035bcba1f73d1941c718608e92b0362aa9d7":{"kind":"REWORK","_number":2,"created":"2021-11-30 22:56:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/819898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/819898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/819898/2"}}},"commit":{"parents":[{"commit":"6bbf61a67ba70726f63ff7e287e0e6ad30e4af06","subject":"Merge \"Rename [coordination] heartbeat to hearbeat_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6bbf61a67ba70726f63ff7e287e0e6ad30e4af06"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 15:43:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 22:56:34.000000000","tz":540},"subject":"Load api-paste.ini from configuration directories first","message":"Load api-paste.ini from configuration directories first\n\nThe api-paste.ini file defines WSGI definition of API. This file is\nregarded as one configuration file and sometimes users need to modify\nthe file, for example to enable an additional middleware.\n\nHowever currently aodh-api by default expects the file is located in\nthe aodh source directory and users should set the [api] paste_config\nparameter in addition to copy and modify the file.\n\nThis change makes the aodh-api process search the api-paste.ini file\nfrom configuration directories automatically, so that users can more\neasily customize the file by putting the modified file in /etc/aodh\n(or any other config directories).\n\nIf the paste_config parameter is defined as a full path then aodh-api\ndirectly loads the file from the specified path. If the file is not\nfound then the default file in the aodh source directory is used.\n\nChange-Id: I4a6194fbf6a5317a523d75e3fb6bd2fd055f5100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f95035bcba1f73d1941c718608e92b0362aa9d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f95035bcba1f73d1941c718608e92b0362aa9d7"}]},"branch":"refs/heads/master"},"249dd2b59926122868625b0d940d650facc781ff":{"kind":"REWORK","_number":3,"created":"2021-11-30 23:45:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/819898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/819898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/819898/3"}}},"commit":{"parents":[{"commit":"6bbf61a67ba70726f63ff7e287e0e6ad30e4af06","subject":"Merge \"Rename [coordination] heartbeat to hearbeat_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6bbf61a67ba70726f63ff7e287e0e6ad30e4af06"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 15:43:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 23:44:55.000000000","tz":540},"subject":"Load api-paste.ini from configuration directories first","message":"Load api-paste.ini from configuration directories first\n\nThe api-paste.ini file defines WSGI definition of API. This file is\nregarded as one configuration file and sometimes users need to modify\nthe file, for example to enable an additional middleware.\n\nHowever currently aodh-api by default expects the file is located in\nthe aodh source directory and users should set the [api] paste_config\nparameter in addition to copy and modify the file.\n\nThis change makes the aodh-api process search the api-paste.ini file\nfrom configuration directories automatically, so that users can more\neasily customize the file by putting the modified file in /etc/aodh\n(or any other config directories).\n\nIf the paste_config parameter is defined as a full path then aodh-api\ndirectly loads the file from the specified path. If the file is not\nfound then the default file in the aodh source directory is used.\n\nChange-Id: I4a6194fbf6a5317a523d75e3fb6bd2fd055f5100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/249dd2b59926122868625b0d940d650facc781ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/249dd2b59926122868625b0d940d650facc781ff"}]},"branch":"refs/heads/master"},"c2196c83d857ae6dde34b7b633be913c039b5a93":{"kind":"REWORK","_number":4,"created":"2021-11-30 23:48:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/819898/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/819898/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/819898/4"}}},"commit":{"parents":[{"commit":"6bbf61a67ba70726f63ff7e287e0e6ad30e4af06","subject":"Merge \"Rename [coordination] heartbeat to hearbeat_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6bbf61a67ba70726f63ff7e287e0e6ad30e4af06"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 15:43:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 23:48:21.000000000","tz":540},"subject":"Load api-paste.ini from configuration directories first","message":"Load api-paste.ini from configuration directories first\n\nThe api-paste.ini file defines WSGI definition of API. This file is\nregarded as one configuration file and sometimes users need to modify\nthe file, for example to enable an additional middleware.\n\nHowever currently aodh-api by default expects the file is located in\nthe aodh source directory and users should set the [api] paste_config\nparameter in addition to copy and modify the file.\n\nThis change makes the aodh-api process search the api-paste.ini file\nfrom configuration directories automatically, so that users can more\neasily customize the file by putting the modified file in /etc/aodh\n(or any other config directories).\n\nIf the paste_config parameter is defined as a full path then aodh-api\ndirectly loads the file from the specified path. If the file is not\nfound then the default file in the aodh source directory is used.\n\nChange-Id: I4a6194fbf6a5317a523d75e3fb6bd2fd055f5100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c2196c83d857ae6dde34b7b633be913c039b5a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c2196c83d857ae6dde34b7b633be913c039b5a93"}]},"branch":"refs/heads/master"},"6241369e10c165f00113d9f7ba53a52a60a67aae":{"kind":"REWORK","_number":5,"created":"2021-12-01 01:35:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/819898/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/819898/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/819898/5"}}},"commit":{"parents":[{"commit":"6bbf61a67ba70726f63ff7e287e0e6ad30e4af06","subject":"Merge \"Rename [coordination] heartbeat to hearbeat_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6bbf61a67ba70726f63ff7e287e0e6ad30e4af06"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 15:43:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-01 01:35:28.000000000","tz":540},"subject":"Load api-paste.ini from configuration directories first","message":"Load api-paste.ini from configuration directories first\n\nThe api-paste.ini file defines WSGI definition of API. This file is\nregarded as one configuration file and sometimes users need to modify\nthe file, for example to enable an additional middleware.\n\nHowever currently aodh-api by default expects the file is located in\nthe aodh source directory and users should set the [api] paste_config\nparameter in addition to copy and modify the file.\n\nThis change makes the aodh-api process search the api-paste.ini file\nfrom configuration directories automatically, so that users can more\neasily customize the file by putting the modified file in /etc/aodh\n(or any other config directories).\n\nIf the paste_config parameter is defined as a full path then aodh-api\ndirectly loads the file from the specified path. If the file is not\nfound then the default file in the aodh source directory is used.\n\nChange-Id: I4a6194fbf6a5317a523d75e3fb6bd2fd055f5100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6241369e10c165f00113d9f7ba53a52a60a67aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6241369e10c165f00113d9f7ba53a52a60a67aae"}]},"branch":"refs/heads/master"},"8ac7c65cffac34e412ab117f8e86b4dd2c973ac9":{"kind":"REWORK","_number":6,"created":"2021-12-01 01:50:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/819898/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/819898/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/819898/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/819898/6"}}},"commit":{"parents":[{"commit":"6bbf61a67ba70726f63ff7e287e0e6ad30e4af06","subject":"Merge \"Rename [coordination] heartbeat to hearbeat_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6bbf61a67ba70726f63ff7e287e0e6ad30e4af06"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 15:43:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-01 01:50:31.000000000","tz":540},"subject":"Load api-paste.ini from configuration directories first","message":"Load api-paste.ini from configuration directories first\n\nThe api-paste.ini file defines WSGI definition of API. This file is\nregarded as one configuration file and sometimes users need to modify\nthe file, for example to enable an additional middleware.\n\nHowever currently aodh-api by default expects the file is located in\nthe aodh source directory and users should set the [api] paste_config\nparameter in addition to copy and modify the file.\n\nThis change makes the aodh-api process search the api-paste.ini file\nfrom configuration directories automatically, so that users can more\neasily customize the file by putting the modified file in /etc/aodh\n(or any other config directories).\n\nIf the paste_config parameter is defined as a full path then aodh-api\ndirectly loads the file from the specified path. If the file is not\nfound then the default file in the aodh source directory is used.\n\nChange-Id: I4a6194fbf6a5317a523d75e3fb6bd2fd055f5100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8ac7c65cffac34e412ab117f8e86b4dd2c973ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8ac7c65cffac34e412ab117f8e86b4dd2c973ac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
