)]}'
{"id":"openstack%2Faodh~822640","triplet_id":"openstack%2Faodh~master~I148b256c0eebb597dedff46cacac13092603e816","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-09-06 05:52:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I148b256c0eebb597dedff46cacac13092603e816","subject":"Remove unicode literal from code","status":"ABANDONED","created":"2021-12-22 09:36:19.000000000","updated":"2022-09-06 05:52:37.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4f5e32a9cb16ef815b68ce79cd5c9e1110bb01ca","_number":822640,"virtual_id_number":822640,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 09:50:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-29 10:33:29.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2021-12-30 01:41:33.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-09-06 05:51:24.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"}],"messages":[{"id":"219b84739c54f636f15ff2cb0741ae5125d700a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2021-12-22 09:36:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48e85540a35579ac94565520e4db7921c42a139b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 09:50:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9e1a0228c34c85930f4001ec5938bb : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740af3d73e7240a09b4b234f9a815025 : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0b184ad890146cfd207b557fa6d570ed3e30c3fe","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-12-29 10:33:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ce0545376bd1973c24022964bb24334144c2e041","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-12-30 01:41:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"624a343c3d63ad0ae4d8e84ccc51cafbe91388ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-09-06 02:01:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"35e536fa1dc55bcda4e53e7c796bfc06ffaa58c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 02:25:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2edb95fb79a849dc8d5f8027f0be49c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9713e3a01c0e4206bbe7817dff6107e6 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa76a5d36e24afdbe45ac4dc6e63a82 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":2},{"id":"379d838cf33d8d7d5a550f99089fe19ea0b99201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-09-06 05:51:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6fde42eec234218b438f6b698309e3ca9a70ffbc","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-09-06 05:51:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4f5e32a9cb16ef815b68ce79cd5c9e1110bb01ca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-09-06 05:52:37.000000000","message":"Abandoned\n\nThank you for the proposal, the change was incorporated in another change fixing the whole unicode stuff.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8ccc90ffcbc379f0c6659d2a93ba45754ecf1884","revisions":{"1f12160e44346b2d75ecbf7428e50e53398ec1ec":{"kind":"REWORK","_number":1,"created":"2021-12-22 09:36:19.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/40/822640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/40/822640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/40/822640/1"}}},"commit":{"parents":[{"commit":"8420354f6e9c09a0024b67cdb99404e7df7d34ad","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8420354f6e9c09a0024b67cdb99404e7df7d34ad"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2021-12-22 09:33:29.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2021-12-22 09:33:29.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I148b256c0eebb597dedff46cacac13092603e816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1f12160e44346b2d75ecbf7428e50e53398ec1ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1f12160e44346b2d75ecbf7428e50e53398ec1ec"}]},"branch":"refs/heads/master"},"7c363f7a2fddd970a31c2fb1fe5c8aff52dc3b1e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-06 02:01:18.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/40/822640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/40/822640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/40/822640/2"}}},"commit":{"parents":[{"commit":"8420354f6e9c09a0024b67cdb99404e7df7d34ad","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8420354f6e9c09a0024b67cdb99404e7df7d34ad"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2021-12-22 09:33:29.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-09-06 02:01:18.000000000","tz":0},"subject":"Remove unicode literal from code","message":"Remove unicode literal from code\n\nAll strings are considered as unicode literal string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I148b256c0eebb597dedff46cacac13092603e816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7c363f7a2fddd970a31c2fb1fe5c8aff52dc3b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7c363f7a2fddd970a31c2fb1fe5c8aff52dc3b1e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8ccc90ffcbc379f0c6659d2a93ba45754ecf1884":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-06 05:51:05.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/40/822640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/40/822640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/822640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/40/822640/3"}}},"commit":{"parents":[{"commit":"165c96b51d64458d11abbc5e93cf8262cb2f5e63","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/165c96b51d64458d11abbc5e93cf8262cb2f5e63"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2021-12-22 09:33:29.000000000","tz":480},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2022-09-06 05:51:05.000000000","tz":0},"subject":"Remove unicode literal from code","message":"Remove unicode literal from code\n\nAll strings are considered as unicode literal string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I148b256c0eebb597dedff46cacac13092603e816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8ccc90ffcbc379f0c6659d2a93ba45754ecf1884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8ccc90ffcbc379f0c6659d2a93ba45754ecf1884"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
