)]}'
{"id":"openstack%2Faodh~824265","triplet_id":"openstack%2Faodh~stable%2Fqueens~I2bab94314077f28589d872076d4116310c61a72a","project":"openstack/aodh","branch":"stable/queens","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-05 18:38:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2bab94314077f28589d872076d4116310c61a72a","subject":"[Queens] Remove TripleO jobs","status":"ABANDONED","created":"2022-01-12 00:56:47.000000000","updated":"2022-12-05 18:38:19.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"54b2d2184ba1fbba09a581c4a87d367698288374","_number":824265,"virtual_id_number":824265,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-12 01:29:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 01:29:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 08:52:59.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"}],"messages":[{"id":"05cf25085dd2a458e249e9040da67003855fca2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 00:56:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"567447c2ff7aa51c18f4d62933eb69ea75037a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 01:29:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/839fc767b2604e94872acadaa722c916 : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de030505955f448ea3b94e8c2d345813 : FAILURE in 5m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3ad2c74a8b4748558db2de3a8a14b6a0 : SUCCESS in 9m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d61941eb43fb4d80a656b8ee40585233 : SUCCESS in 5m 57s\n- aodh-dsvm-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/7ae21175cc9243929db878ad990eba9f : FAILURE in 25m 43s\n- aodh-dsvm-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/956c5f4f51fc4c048897b76507043d6e : FAILURE in 26m 11s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e9bd863fa6474eb495e73599bd605685 : FAILURE in 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7db53d75e44aa1992dbfbe8ed0b71227954f1c2f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:52:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b3092f5fd4349742fa6475ecd2d38218b7da4d0","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:56:36.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"54b2d2184ba1fbba09a581c4a87d367698288374","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-05 18:38:19.000000000","message":"Abandoned\n\nThis branch of the project has been moved to End of Life. In order to be able to delete the branch open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20b522b4895738cd95b5ede79bef429a04e1ab08","revisions":{"20b522b4895738cd95b5ede79bef429a04e1ab08":{"kind":"REWORK","_number":1,"created":"2022-01-12 00:56:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/824265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/65/824265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/65/824265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/65/824265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/65/824265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/65/824265/1"}}},"commit":{"parents":[{"commit":"fce6e7b430c4732425e2ab41ec2bd231abda4840","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fce6e7b430c4732425e2ab41ec2bd231abda4840"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 00:56:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 00:56:07.000000000","tz":540},"subject":"[Queens] Remove TripleO jobs","message":"[Queens] Remove TripleO jobs\n\nTripleO has retired its stable/queens, and the branch is no longer\nmaintained.\n\nChange-Id: I2bab94314077f28589d872076d4116310c61a72a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/20b522b4895738cd95b5ede79bef429a04e1ab08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/20b522b4895738cd95b5ede79bef429a04e1ab08"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
