)]}'
{"id":"openstack%2Faodh~824266","triplet_id":"openstack%2Faodh~stable%2Frocky~I7c81b696478e7af1ca6f98193a2f09d4d03e6d8a","project":"openstack/aodh","branch":"stable/rocky","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-05-23 17:57:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7c81b696478e7af1ca6f98193a2f09d4d03e6d8a","subject":"[Rocky] Remove TripleO jobs","status":"ABANDONED","created":"2022-01-12 00:57:43.000000000","updated":"2023-05-23 17:57:28.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"471f011aa7ac7db63dd6f8d846be0668069aa0db","_number":824266,"virtual_id_number":824266,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-12 01:35:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 01:35:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 08:53:23.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"}],"messages":[{"id":"12e9b2a37ad225efc0168643467111da8584a27e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 00:57:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5110c9cf46d7157003e017c9f4d70cd348e3ca6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 01:35:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7b015a7efc46d088370425955d6d99 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5731bb407854b4894c23685a67149c4 : FAILURE in 6m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3172700a555447818e57c702f8a905e0 : SUCCESS in 8m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bfc6178e18e44ed1b0beb958f46958eb : SUCCESS in 5m 30s\n- aodh-dsvm-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/322cbcccd0e648169982a9754e790e80 : FAILURE in 23m 33s\n- aodh-dsvm-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7e41a2d13022455e9c5fe71e1589fc2b : FAILURE in 23m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6bf363f2b02548d2923fdb580373ee59 : FAILURE in 32m 24s","accounts_in_message":[],"_revision_number":1},{"id":"467debf67fb14293bbea25749f346d127f496b44","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:53:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"471f011aa7ac7db63dd6f8d846be0668069aa0db","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-05-23 17:57:28.000000000","message":"Abandoned\n\nrocky branch of this project is marked as End of Life. Open patches need to be abandoned in order to be able to delete the branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d9ceef45e5f57dd49c60c581169cc543772b17a","revisions":{"8d9ceef45e5f57dd49c60c581169cc543772b17a":{"kind":"REWORK","_number":1,"created":"2022-01-12 00:57:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/824266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/66/824266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/824266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/824266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/824266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/66/824266/1"}}},"commit":{"parents":[{"commit":"887a4953ac82667ee77efca98527fa6b3c151ed5","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/887a4953ac82667ee77efca98527fa6b3c151ed5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 00:57:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 00:57:14.000000000","tz":540},"subject":"[Rocky] Remove TripleO jobs","message":"[Rocky] Remove TripleO jobs\n\nTripleO has retired its stable/rocky, and the branch is no longer\nmaintained.\n\nChange-Id: I7c81b696478e7af1ca6f98193a2f09d4d03e6d8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8d9ceef45e5f57dd49c60c581169cc543772b17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8d9ceef45e5f57dd49c60c581169cc543772b17a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
