)]}'
{"id":"openstack%2Faodh~824267","triplet_id":"openstack%2Faodh~stable%2Fstein~If208e9d68ef7aed91e6af3a1fe06f06543246b30","project":"openstack/aodh","branch":"stable/stein","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-07 14:33:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If208e9d68ef7aed91e6af3a1fe06f06543246b30","subject":"[Stein] Remove TripleO jobs","status":"ABANDONED","created":"2022-01-12 00:58:24.000000000","updated":"2023-09-07 14:33:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9638c2329d9a4822a98ae475c925f936a1c17376","_number":824267,"virtual_id_number":824267,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-12 01:29:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 01:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 08:57:04.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"}],"messages":[{"id":"78436c7806c02c9600a6f6b81f7eb95767332c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 00:58:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"834d6ee71f3ceb9e2bd7112b8d049c86075a2bab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 01:29:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af7dd7ecf0849f0924f90376dbbc8e1 : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36016b408da2434f94b27ecf6d30c5c7 : FAILURE in 6m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3c793beb89044aa5932ce2444979a0a3 : SUCCESS in 12m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/094b97c5cefd4bb6a82a4d9699f9332c : FAILURE in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/290b0f0b0d174b88a7784a5b54bff3c5 : SUCCESS in 4m 45s\n- aodh-dsvm-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a20b9b6e05da4a24b33d920e0c5df458 : FAILURE in 21m 02s\n- aodh-dsvm-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/9e2fd8088310455898ea81018b70bd13 : FAILURE in 25m 50s (non-voting)\n- telemetry-tox-py37 https://zuul.opendev.org/t/openstack/build/59b6ad22c14e4e35a433a6f2ca1a56be : FAILURE in 9m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c7f448c2d6711c5e5c9ecdcfa27276cf0f4ba757","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:57:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9638c2329d9a4822a98ae475c925f936a1c17376","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-07 14:33:01.000000000","message":"Abandoned\n\nThis project has transitioned its stable/stein branch to End of Life. All open patches need to be closed to be able to delete the branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60be46c8cfd44495b1966a9677cabeb55c8d7702","revisions":{"60be46c8cfd44495b1966a9677cabeb55c8d7702":{"kind":"REWORK","_number":1,"created":"2022-01-12 00:58:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/824267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/67/824267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/67/824267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/67/824267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/67/824267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/67/824267/1"}}},"commit":{"parents":[{"commit":"43e3b39a649897344ca863e6c8b4e85d5b32d4b7","subject":"Merge \"Byte encoding of member Id\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/43e3b39a649897344ca863e6c8b4e85d5b32d4b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 00:58:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 00:58:02.000000000","tz":540},"subject":"[Stein] Remove TripleO jobs","message":"[Stein] Remove TripleO jobs\n\nTripleO has retired its stable/stein, and the branch is no longer\nmaintained.\n\nChange-Id: If208e9d68ef7aed91e6af3a1fe06f06543246b30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/60be46c8cfd44495b1966a9677cabeb55c8d7702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/60be46c8cfd44495b1966a9677cabeb55c8d7702"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
