)]}'
{"id":"openstack%2Faodh~824268","triplet_id":"openstack%2Faodh~stable%2Fxena~I063097b637134cc8d623082e1cf0c49afc92e379","project":"openstack/aodh","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-16 15:04:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I063097b637134cc8d623082e1cf0c49afc92e379","subject":"[Xena] Remove TripleO job","status":"MERGED","created":"2022-01-12 00:59:30.000000000","updated":"2022-02-16 15:09:41.000000000","submitted":"2022-02-16 15:04:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824268","meta_rev_id":"2635d48bf7010455ae35964f69af98aaf7c31612","_number":824268,"virtual_id_number":824268,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 15:04:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-02-16 13:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-02-16 13:58:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 02:05:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 08:33:44.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2022-02-16 13:58:41.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"5ef669081545a738bd5cca884375864babd441e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 00:59:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d26b8571c2a02f208c0184d68431a04111ce0e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 02:05:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6c71813b574b3e8511ca44dff269ef : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d7a4f9648cf4961949b9b5b03852a4b : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/800f32eba25441ae861b0e0a29f0d05f : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57d29733448447d5a0af26c893ade38f : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c3eab733574e25a997ab8aa01c6dfa : SUCCESS in 7m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/afead99f01854569882dd3ef8e460992 : FAILURE in 58m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/afc8a024b7844cbfb0ffddeda6415ddf : SUCCESS in 42m 27s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bd681d73b9c3487aaf7bece3b72e1ecd : SUCCESS in 45m 26s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/99b86f49a8304e6990e7afd14eb7b6c1 : SUCCESS in 45m 08s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/64b7a1d9314d475287f70ad967a64831 : SUCCESS in 49m 22s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/9d5ee40ba9cb484dbadb706d8b23cfdd : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0e95b4dd15b1ce8c93477886e70d666c0608ff6b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-01-12 08:33:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2661bc8f6961fcdd48f20bc45dc01cc311166f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 09:45:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d69552bffdf4c43aac26ae1f238e260 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/089ea02c14b1407d8d1d80437725ea78 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1929023c08a5442a8985cbe2d2da97d4 : SUCCESS in 12m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/794c901c104b4153953ebe244195a3bd : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d67ccec7ea954ee48f2632d34eaa40ff : SUCCESS in 11m 55s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b7ae253215fc43769d1526e5fc4f83d4 : SUCCESS in 47m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7c492e58479e44b0a3b6fcc9eefe3730 : SUCCESS in 44m 56s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f7d612f15c5d4463906335ffd02e84bd : SUCCESS in 52m 13s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/cdd623587f5f4084a0505f205c84a550 : SUCCESS in 54m 27s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/07624d38d005435f99ed5ac34b7a7210 : SUCCESS in 53m 58s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/c17c5225ee94409481ac331b07fb845f : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"32895ea16d21c7b87f809e2375b694af5513b8bf","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-02-16 13:58:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"308bc6c0a6e08975a0eb4d4d66ed6686de5961c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 13:59:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df5951c2bad3a7d4b87b91ed8f8dacbae278b496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:04:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/146cd47c0d3b4cf19e9874ae7690bcd1 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eb74b7002b44a62b6e59fce9441598d : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d027cd06d979409c9c008f03b8b4a5a5 : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de61e1b6d0a468f86f9618d38919fab : SUCCESS in 8m 13s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4dc3ef9b4fa0474da8693adb5346f610 : SUCCESS in 42m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ef2291582e90476eb6f0ac20b1e5abf8 : SUCCESS in 41m 44s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/90a6b528725a478a8b8d6230713569e0 : SUCCESS in 44m 18s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/cf4f6f99e67b4572ae4e8aaef262829b : SUCCESS in 44m 57s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/64b48352415d4063b9608a53c9848c41 : SUCCESS in 58m 12s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/3ce1c5a5b18645179bafa0f082bb50b6 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5d66f00df1ec04cc97c7d32487afc4219f02ca94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:04:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2635d48bf7010455ae35964f69af98aaf7c31612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:09:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906314e717d0477d8d654968e4b62670 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8789bc591fafb8bffd58a358143eadd7d353f35e","revisions":{"8789bc591fafb8bffd58a358143eadd7d353f35e":{"kind":"REWORK","_number":1,"created":"2022-01-12 00:59:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/824268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/68/824268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/824268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/824268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/824268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/68/824268/1"}}},"commit":{"parents":[{"commit":"5ac3421cf49cc9ecb972ae856200fce934c54a42","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5ac3421cf49cc9ecb972ae856200fce934c54a42"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 00:58:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 00:58:51.000000000","tz":540},"subject":"[Xena] Remove TripleO job","message":"[Xena] Remove TripleO job\n\nTripleO has transitioned into the independent release model and doesn\u0027t\ncreate its stable/xena.\n\nChange-Id: I063097b637134cc8d623082e1cf0c49afc92e379\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8789bc591fafb8bffd58a358143eadd7d353f35e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8789bc591fafb8bffd58a358143eadd7d353f35e"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
