)]}'
{"id":"openstack%2Faodh~832561","triplet_id":"openstack%2Faodh~master~I09d998595f20ce36c61972e6f8b0f2e9dac68def","project":"openstack/aodh","branch":"master","topic":"add-zed-python-jobtemplates","hashtags":[],"change_id":"I09d998595f20ce36c61972e6f8b0f2e9dac68def","subject":"Add Python3 zed unit tests","status":"MERGED","created":"2022-03-08 12:05:33.000000000","updated":"2022-03-08 14:59:39.000000000","submitted":"2022-03-08 14:54:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832561-add-zed-python-jobtemplates","meta_rev_id":"bdead4ba84f791523135762c867010f2bc5147a0","_number":832561,"virtual_id_number":832561,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 14:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-03-08 13:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-03-08 13:35:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 13:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-08 13:35:37.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"20354e5cd6d37d38329152a8307c8c5c399eda03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-03-08 12:05:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ea531f59b56df10f4785ebd0046192fb212233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 13:16:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902e556fe0d24a98b5c03b066ea248cf : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c948ac1c29784116a686d47500421cce : SUCCESS in 10m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05c234ab3cc2447184477b51e5cf44c1 : SUCCESS in 11m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c05b9a8e387414eba106f388485f4f5 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8630c44f42184105ad09f48aff78a1a6 : SUCCESS in 10m 31s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a4d9ef3bea7a459aa00f984e87727ffd : SUCCESS in 36m 10s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/67c0f82190cd432081df94ca83c63313 : SUCCESS in 42m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/20f6ac4e5f31483ea084402acd9a1d62 : SUCCESS in 35m 52s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/13bc12ad532747918ff5f7a81532c44a : SUCCESS in 39m 51s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/028e4c7d4d4542b7983be5ba53e7f86b : SUCCESS in 52m 00s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/75ea271227514779b1ac1edc3e0c3b7a : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9e6ee236a805a8e9cc4caa0dd48f88b83a245df5","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-03-08 13:35:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3f42bec8d52e42a42f93e2e0b7a2fad1da4fb73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 13:36:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08c02a2201d72e2e4677da29ad08b0c38cbe8137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:54:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abfb5f3fa81f48da809fa590ee713be1 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50cf664e890b4f7f883fa2edcf4096a3 : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/defb70ebeb1a48c4ac3c306708e9c9f7 : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c2c3851e134af29936830b9dcb3635 : SUCCESS in 8m 58s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3eeab461d8494f9a86f3ac88d0d1563b : SUCCESS in 46m 00s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/25ed738f21a54d97b1bca155ac881a37 : SUCCESS in 46m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b50a1dc7c51c4720b7caa0a18f28ee19 : SUCCESS in 55m 17s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/773290a18e2b494799fbbd75d7b09857 : SUCCESS in 54m 44s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/7e5d4be23edd459d99a8dda808718a6d : SUCCESS in 1h 11m 25s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/b4a413076d104aec9104186536df54c4 : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b8025b04d5bc75d8009922deca4834430e738ec0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:54:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bdead4ba84f791523135762c867010f2bc5147a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:59:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b2e5998a1245b995e84e6a5b78eeef : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11eb43d68e502473e64631164bc6c101289d64bc","revisions":{"11eb43d68e502473e64631164bc6c101289d64bc":{"kind":"REWORK","_number":1,"created":"2022-03-08 12:05:33.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/61/832561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/61/832561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/61/832561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/61/832561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/61/832561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/61/832561/1"}}},"commit":{"parents":[{"commit":"190c9e1d217f5b6f5a52ae5d975d928066cbe750","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/190c9e1d217f5b6f5a52ae5d975d928066cbe750"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-08 12:05:31.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-08 12:05:31.000000000","tz":0},"subject":"Add Python3 zed unit tests","message":"Add Python3 zed unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for zed.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I09d998595f20ce36c61972e6f8b0f2e9dac68def\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/11eb43d68e502473e64631164bc6c101289d64bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/11eb43d68e502473e64631164bc6c101289d64bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
