)]}'
{"id":"openstack%2Faodh~833122","triplet_id":"openstack%2Faodh~stable%2Fwallaby~I4cfcb3e12012075eb56c6f69200825f0b597803f","project":"openstack/aodh","branch":"stable/wallaby","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"30893":{"account":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"last_update":"2022-03-17 08:25:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4cfcb3e12012075eb56c6f69200825f0b597803f","subject":"Adding FIPS job","status":"MERGED","created":"2022-03-10 16:25:57.000000000","updated":"2022-03-17 08:30:25.000000000","submitted":"2022-03-17 08:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833122-fips-compatibility","meta_rev_id":"e2a45bd6d3f87eb1101338f57293a03c4fc5457c","_number":833122,"virtual_id_number":833122,"owner":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-17 08:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-03-17 07:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-10 16:30:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-03-17 07:05:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 16:30:21.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"state":"REVIEWER"},{"updated":"2022-03-10 18:00:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 18:32:10.000000000","updated_by":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"CC"},{"updated":"2022-03-10 18:32:10.000000000","updated_by":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2022-03-17 07:05:00.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"b03ff82ee5378e68d64ac856c8863ff27a90eb2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2022-03-10 16:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b61436213a3c9bf4c559c42259331230e884e4fe","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2022-03-10 16:30:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b30925879409f334945a645b4d4a6cd88be16375","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2022-03-10 17:11:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c08de18480288e893d5b575a94fe9e7eb03cec80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 18:00:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/224a0022e9f74093ad33f6ca22f267ed : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/760ee77ace6d447589595de0c9a35be0 : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e46f46c6fa240d98ae35b7f7150ed97 : SUCCESS in 10m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92f6100d6338474bb3af30989c70e41b : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7609341a7f8d4fdf8763cd88cbab655f : SUCCESS in 9m 23s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0b3d083cf54e4567926264bd45d38dbc : SUCCESS in 55m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d998674783754d6982b48cb689bf2f97 : SUCCESS in 53m 48s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9aea7bcd02ff4daf95dd1d4f31a26a8c : SUCCESS in 56m 59s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/865711e92a274e6998e6187e716954b9 : SUCCESS in 1h 14m 04s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/51d4876feade4567adcb2edb9fe681f0 : SUCCESS in 1h 05m 44s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/72cc48b68bff472bb0776f6f8dd34897 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":1},{"id":"46324b9f318dbaa45dd9c8e7081502e64a1eff88","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-03-17 07:05:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68890e28ed5266e7f0dd83ab6371b01783feba8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 07:06:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43fea9ea253173cd41c1f6e4afad7501607006dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 08:25:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b411e51b88e4531be2eb876895f68a2 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a03e7bead404429a24a64927a1297f7 : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2fd4ad26b6047a99236571d007221e9 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8be2863f54346208420fa2033a817ef : SUCCESS in 9m 58s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a3d93164b96f44d486b08f829fb32cce : SUCCESS in 54m 56s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/271f581dadd64973bc673c50263df166 : SUCCESS in 55m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2c5b0da5b43e49db9b16a060410f816a : SUCCESS in 58m 53s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/49a14850ee9d463da76fdd4411edfbfc : SUCCESS in 1h 02m 37s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/b73ccd901e8141f49c86e91051e0df87 : SUCCESS in 57m 48s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/fb19407ee88c44f59797a9c1718b8bd1 : SUCCESS in 1h 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3c1a4f3f26a63c27d676d91c7caab2d88ba25f0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 08:25:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e2a45bd6d3f87eb1101338f57293a03c4fc5457c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 08:30:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3454d086efe549c6a9a2ee14a3825ba1 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8f1746e03c121063bdb954dd83d3e62afe3e038","revisions":{"c8f1746e03c121063bdb954dd83d3e62afe3e038":{"kind":"REWORK","_number":1,"created":"2022-03-10 16:25:57.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/22/833122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/833122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/833122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/833122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/833122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/833122/1"}}},"commit":{"parents":[{"commit":"1fb8cde66b81c6e138749956824c7676ba6073a7","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1fb8cde66b81c6e138749956824c7676ba6073a7"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 18:46:19.000000000","tz":-240},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2022-03-10 16:24:26.000000000","tz":-300},"subject":"Adding FIPS job","message":"Adding FIPS job\n\nChange-Id: I4cfcb3e12012075eb56c6f69200825f0b597803f\nDepends-On: I05713ed77b79e583ad1073ff80ccd7a031e34c26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c8f1746e03c121063bdb954dd83d3e62afe3e038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c8f1746e03c121063bdb954dd83d3e62afe3e038"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
