)]}'
{"id":"openstack%2Faodh~837053","triplet_id":"openstack%2Faodh~stable%2Fxena~I4f631d224404460138f4050b1b981d577b592544","project":"openstack/aodh","branch":"stable/xena","topic":"bug/1946793-stable/xena","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2022-11-07 18:30:11.000000000","reason":"Change was submitted"},"5112":{"account":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"last_update":"2022-11-07 18:30:11.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-20 04:35:58.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-09-29 06:15:44.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"10366":{"account":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"last_update":"2022-11-07 18:30:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f631d224404460138f4050b1b981d577b592544","subject":"gnocchi: Use Dynamic Aggregates API","status":"MERGED","created":"2022-04-08 03:41:20.000000000","updated":"2022-11-07 18:31:14.000000000","submitted":"2022-11-07 18:30:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"837053-bug/1946793-stable/xena","cherry_pick_of_change":829870,"cherry_pick_of_patch_set":7,"meta_rev_id":"09737f10f67b07ebc9ee5f9d8dd9cbe983bb8fed","_number":837053,"virtual_id_number":837053,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 18:30:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-11-07 17:15:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-11-07 17:15:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 03:41:20.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"CC"},{"updated":"2022-04-08 04:54:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-11 03:24:00.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2022-04-18 04:00:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-04-20 07:30:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2022-05-21 00:10:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-06-23 08:13:11.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"state":"CC"},{"updated":"2022-10-20 04:17:51.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"CC"}],"messages":[{"id":"e2151ce0e085aae373434f453b36ae181db0b5d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-04-08 03:41:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ce14db66538686c81cc66ba9f09b3694f446cc42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 04:54:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2408915eef43b4ae8b26be8e5adc25 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1109467d0ea2477a91bbb18c7afdd6bc : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09ff87ae088149d7b2677d9817b52278 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42b8c91154954af4a75e312d40bdb673 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b628b0bd53486ea0c79ff5ef1509d3 : SUCCESS in 6m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2664d62eeb7a4f7eb8700a4936fa3ef8 : SUCCESS in 46m 32s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f9188a37631c4c18a3cb56ee4e508dff : SUCCESS in 50m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3eff9838a35c49298ffd086266e440fc : SUCCESS in 56m 17s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/96e91f3f4f324bf882c037c289e5897b : SUCCESS in 50m 08s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/094b0ac9bc0249399c0c252d130429bf : SUCCESS in 1h 02m 27s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/438ba628a8514e4a8574d59f7129074e : SUCCESS in 1h 07m 06s","accounts_in_message":[],"_revision_number":1},{"id":"35cf91d1047c2fccf62b25c1cb008515f9b966de","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-04-11 03:24:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb921b0311647ba87999f323abe863f787d6fa44","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-18 04:00:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9228bc33918f772d764bcc2f16283789003ed47","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-04-19 06:10:05.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4765365a5286e870b1dc5c146b040b98bddec43a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-20 07:30:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"679522013bac198480070017328703c9d4323ab7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 00:10:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53c59521e35b503769a9c2ba95622e15e5516511","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-05-30 04:06:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7af4cbf4964c38e15b66a517f85f308c9f6ca090","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-06-23 08:13:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29516e2a8ea27343daaba1cfd35acd3ae7bbc18d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 09:07:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11fb1fcfbca84e149b71bac27ba961cd : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/489894e9fca544cea549fd8f65fdd63f : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2f07f725ee446dc963b8d5f1cfe6bbc : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c96c5a0065eb495fae2106b86b482b7c : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c57fe2924be44ad9e6ba40337b42be2 : SUCCESS in 7m 46s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a0e8b7952fe04ef2a0dfc59edf9c4534 : SUCCESS in 46m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/79c3445d42c547289c14fba8002a3363 : SUCCESS in 30m 59s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/dd236ba94269480591936a7ae12796ea : SUCCESS in 49m 29s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/270692d4f6204a04890f1fb3311bfad0 : SUCCESS in 47m 18s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/ec499721f44a481ca887886c2637eb37 : FAILURE in 50m 33s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/b6f39c773b8f4e4e868ee255d1d137a1 : RETRY_LIMIT in 9m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a555605ca7682cc66825b6d9b6825b5d3a2a77c6","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-24 14:00:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"146da7976265bf407f13a67abff0e4acd87acd8d","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-07-06 04:18:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"656353b9e5c67af130b11965f2c5c91cfb332b13","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-07-06 05:47:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aedbdf48b71f3644af7eed35f34b56aa9fa45a6a","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-07-11 18:29:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97fd1de874ef6bed393029195ae4a02764e8f592","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 18:52:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"975e4f01acde7494d4a7571b8e76e392bf7474bd","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-09-07 03:16:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb1dc6c4153e64bc6ce8e14f10d88babf212b92c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 04:07:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da511159dcd845ed86147371f28f8e18\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568037d0633d4442be3bd5c8d26d9cc5 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5cc407f8698488e907fcd5f9ada2106 : SUCCESS in 9m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a325276de0a46deb747bd50861006c6 : SUCCESS in 9m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daaa156aa67b4312ae3ec193e6cfc24a : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9282dfab5c6457587e739d6360178d7 : SUCCESS in 7m 30s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/694fb0bd55924fa08bc6bf5cdcd3f307 : SUCCESS in 35m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/036566b83ed44ea78ce6e5b90e558558 : SUCCESS in 50m 16s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bb42208ba541427abb661535f5542538 : SUCCESS in 37m 15s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/eb7137fd2dc24a54b4a48a3e923bea91 : SUCCESS in 35m 53s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/b0f4fbc8e50c46118c09e54b5b3a8e9f : FAILURE in 36m 32s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/86ebd3ae35084c42892f98fd3b07a863 : FAILURE in 45m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0287aa7fa32127f6406aae3532b4a811f88a75e0","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-09-07 07:57:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0ed0d7b7e146cc628d7ce108fb802b892ca3d25","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-09-07 08:02:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30723b0a87b1c54baf8478be799a6f99caa8c9e6","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-09-07 09:42:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2799975cd6e35d76dfbca8979a52240f85fd12d4","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-09-28 00:19:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cac0374e9f54efe34711625b6378140496d1764f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 01:27:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eac26f156864476a9d2609d314310494\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/def7ce85c1334bc4868fdd6d796a76db : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2d01b1f2b0c482bb26928c288e81af1 : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e285fe3de9564d7b8c5dc2db04a3f0c6 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/982843f012884170bf67d217532552c8 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa16034d2b9f4b46a2d3b9fd7baf928d : SUCCESS in 9m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/bf2d4545087f4c23ba69305821e39be5 : SUCCESS in 51m 51s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/06029567ab3a4388bce3bb33a9702896 : SUCCESS in 48m 42s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f71175690ead4367bbf204d9d68a9448 : SUCCESS in 55m 51s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/bb4c58cd467b4e51bfacae7697703a60 : SUCCESS in 53m 51s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/e7c14b36164f4d2dbd9399303b03daef : FAILURE in 54m 02s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/faf1537fd09d4605acef42738618dbd0 : FAILURE in 56m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1397dcd0ab2dacb95af5a7344921eeee3383926d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-09-29 06:15:44.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ab8acf2f137afa60f058f63ebcdb63d9d5b5e829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 07:25:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e68ca0af5aae4ee3b014f7f0bd9b29ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b840d44944aa4938bcab4fc61469d4de : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28ada367bb0347c5ab57881843ab7fd5 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4ae4d1c58664bbb80aec716f1b537bb : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/502be2ffcd0b4c7a9f39218a86419a93 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b124b0c7b14c54b513d88d4c856e1d : SUCCESS in 8m 11s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/27e7b41ca7a2470e8a5303cb6ae546ed : SUCCESS in 41m 53s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/30c7e6933ecb48f38458f6ba8f7cb353 : SUCCESS in 33m 28s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ab90b20e94f2465ea7a79e612280f29a : SUCCESS in 47m 16s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/781048a0844e4427ae8f9d2f456ac41f : SUCCESS in 51m 43s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/c1e0fa29f43e4e47a1aed2f9ed0306e6 : SUCCESS in 1h 08m 35s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/584f18632d3e404b9993976f17b91b67 : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":1},{"id":"214462c2b143b524fe4a477f6af672588d000d5a","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-10-13 02:43:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb658adfb371df1e63edc7efcbb77c8817445937","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-13 06:42:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30c1a476f3eb3ef960551d816f8cc313e4aa3239","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-13 06:44:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc6a971ee630ae8f7139f50671358c12df1ffcef","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-19 17:36:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c64dc8a84c0964c10e0b4b4c2b69452077f4885b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 04:35:58.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"359152c85f38f2ab3c6f892b75db1364d3f011f0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 04:36:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ba6b33282487d2fa80a16cea335f3eff0cb2b4c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-26 14:46:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"143fcd0b4e8e4db013ccdec3f806f114f13c4974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 14:47:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e8d8f506151734d581673efecc5ef2a19fa4fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 16:00:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e556d9c0148c4e918518f2f8de75c037\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b13a1010aa84f608ff25f8a1f1654e0 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b5ba638651f4d49a13df46763270f50 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbb38228c82e4c4abbb0bfcc136f9ecd : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55cf13d51e08407892fdc46ed174325f : SUCCESS in 6m 54s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c57e9ebad4c84f25a08c90d185d2e8bc : FAILURE in 37m 38s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8c1a7cca4370479580b6ab89de0933e6 : FAILURE in 35m 52s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2ec71586dc7042aab52d0919a2b179ce : FAILURE in 40m 11s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/47b7b27ac153456a9148d71b9432703b : FAILURE in 42m 57s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/5de0a57cb428408a84277b78ec8bc4e4 : SUCCESS in 1h 01m 23s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/04994e5405534f43966aee0b738c3e61 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a586316fb2d2026e4925456483f5caab106c1458","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-10-27 04:37:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46dd45a590f72cd9849514e132632d179e725b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 05:48:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c472491a343a444eb5f78721cc762db7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1e96c0c2ed4cd4b703495586646f25 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/453b49a734c64ba48b7a7346b799f32d : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4fd8ca382e843bb9997d3c2c5f81835 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4dab39f5aa44515a6a76c8ceb039d95 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629c8a48528c44218cca3c71d3dc6d05 : SUCCESS in 6m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6034869a3f7845ab8a3c4da3c2104511 : FAILURE in 38m 31s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b42c8d682bcb427b8fa59c26af8db3a8 : FAILURE in 39m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f7693d18ad55452d8a0ad8b79cd77ac4 : FAILURE in 39m 54s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c52ac4b6e01f467680f4c9b7f054dd8f : FAILURE in 34m 22s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/7ac4f5050f2f460e8b24fd52987fb2b0 : SUCCESS in 1h 01m 42s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/9ad5ddf184d04607814eeb56dbaaccb8 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7fdf65e1583c1c62f7ee01d47999ed464304e9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-10-31 06:48:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4659edf0f67dc3c6dea8e154c1221437b0cc34ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 07:41:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdce6c3ef23a47659b6478dfc4628e39\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641a304e383d4b719963f2c834cf55ec : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a091674184ed42f6905f387fcf70a2aa : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6f4fb1c65a742e8893554ed51e929b1 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85e15fffbbcd4e0eb4d98caedb2e96a8 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2f54adab014891b632add0a9a8e4aa : SUCCESS in 7m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3a6a6bd360d444849d8239e5a1dcab25 : FAILURE in 23m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d2d5d4c9de594234b2ca362bf9feb02f : FAILURE in 38m 47s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6a84569b33c24628beed70ba40df9f85 : FAILURE in 22m 25s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b93998d865da4718a53de66a2beaea36 : FAILURE in 23m 54s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/a6a93d8e34524e768f7beef1d2bd19f4 : SUCCESS in 33m 51s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/c763fe46294c4883ac00b8a9256a5ab4 : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":2},{"id":"32c96ad6b8f919b89443ae738f2f793774250a21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-07 15:38:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"045d978ab29a7e7e2d193d6bc6b70a870beb4705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:59:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dde6a991b6dc4147bca3904cef10fcd1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3013c6d071c478db54ea2633b141f13 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20674cb5270a409dad16737b2ade39db : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95ebb11587c94fea9b98e1d6036ab9ea : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7aea27ebf36c4269827495ecdbeb8412 : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28794b7e45e0480fa19cc82ae81a6004 : SUCCESS in 6m 23s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/61f489fde1cd4a88bfe7bfcdb97dafee : SUCCESS in 1h 04m 57s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4fe7eb1802b54154ad711762d719134e : SUCCESS in 47m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b3792869050d4a5a8823f07c876e43b0 : SUCCESS in 40m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/43eabbcde1ec4e49ab1b3348c2fd4c7e : SUCCESS in 45m 16s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/0b306ab9be3b413b8fa3b729490e22c7 : SUCCESS in 47m 22s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/69596e1523c24d20ac6ef09d43e8bcca : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":3},{"id":"56311b913e6dd826ae3e2730c2e37ef436bfb2c1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-07 17:15:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"42570f3322f666e6446900bf1efc8ecd181dfe08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 17:16:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d47103f2d2cca04f8ea9eda1622108df69a475ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:30:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f42061b055cc4b778348ba10d33a1873\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ba934b0719490ea7ecdfa75369c3e2 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/145be59d6afb49358fca2704dc597651 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782f1d8fd1164ddaaad2b294794760e1 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ae3cfd4666418bad1e05055038414f : SUCCESS in 6m 06s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d5cb8dd6980841429fbfb9de2ade005a : SUCCESS in 49m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ccee07f53bef40ff9cd0154dd54e2c85 : SUCCESS in 26m 35s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b9fac363f6d44316b2d8df0d229c4f1f : SUCCESS in 40m 58s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6ba16d57fdfc4378ac099696ddc339e5 : SUCCESS in 28m 57s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/0e9462ff17a9498b8f9baf6443397f65 : SUCCESS in 54m 52s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/253bccf2c73e4b00a926583307fe6bb6 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":3},{"id":"3827cea810ace9cee09a1ea6e719e811f8148446","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:30:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"09737f10f67b07ebc9ee5f9d8dd9cbe983bb8fed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:31:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a786e0beded4c498c7cee04d4297bf4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd6c4fec43947d6b0fc26d5bfe40f0d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e6d55b1ea9ee852cc56002c865c9443b526f2f58","revisions":{"9d95ddec6cbe680c79f19d1c013d5db360e4303e":{"kind":"REWORK","_number":1,"created":"2022-04-08 03:41:20.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/53/837053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/53/837053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/53/837053/1"}}},"commit":{"parents":[{"commit":"8789bc591fafb8bffd58a358143eadd7d353f35e","subject":"[Xena] Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8789bc591fafb8bffd58a358143eadd7d353f35e"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2022-04-08 03:41:20.000000000","tz":0},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9d95ddec6cbe680c79f19d1c013d5db360e4303e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9d95ddec6cbe680c79f19d1c013d5db360e4303e"}]},"branch":"refs/heads/stable/xena"},"508ca6001fd34821ca4dce7b6da4cf45fcd4ca26":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-31 06:48:30.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/53/837053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/53/837053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/53/837053/2"}}},"commit":{"parents":[{"commit":"231f404ad13bac69964f19b9b1bfa5a42c1f977a","subject":"Ignore Gnocchi API error when the metric is not yet created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/231f404ad13bac69964f19b9b1bfa5a42c1f977a"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-10-31 06:48:30.000000000","tz":0},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/508ca6001fd34821ca4dce7b6da4cf45fcd4ca26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/508ca6001fd34821ca4dce7b6da4cf45fcd4ca26"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"e6d55b1ea9ee852cc56002c865c9443b526f2f58":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-07 15:38:36.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/53/837053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/53/837053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/53/837053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/53/837053/3"}}},"commit":{"parents":[{"commit":"231f404ad13bac69964f19b9b1bfa5a42c1f977a","subject":"Ignore Gnocchi API error when the metric is not yet created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/231f404ad13bac69964f19b9b1bfa5a42c1f977a"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-07 15:38:36.000000000","tz":0},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\nDepends-On: https://review.opendev.org/863284\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e6d55b1ea9ee852cc56002c865c9443b526f2f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e6d55b1ea9ee852cc56002c865c9443b526f2f58"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
