)]}'
{"id":"openstack%2Faodh~837864","triplet_id":"openstack%2Faodh~master~I6a78a749c2b2f03195dae3390c54663632cc8069","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-13 07:55:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a78a749c2b2f03195dae3390c54663632cc8069","subject":"Migrate evaluation_interval to [evaluator]","status":"MERGED","created":"2022-04-14 12:51:21.000000000","updated":"2022-06-13 07:57:09.000000000","submitted":"2022-06-13 07:55:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837864","meta_rev_id":"ea971a601a293341db33e4b78429c0bc9eeba555","_number":837864,"virtual_id_number":837864,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-13 07:55:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-04-19 07:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2022-04-21 11:05:22.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2022-04-21 11:05:22.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-14 13:45:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 07:23:14.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2022-04-21 11:05:22.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"8d513c5338051c9bbdc796335b240cebf8c58bd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-14 12:51:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bb0d4c7395ea2a0b935e48f7ee4b275a523f0c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 13:45:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc4462226d64a4aa95dec4abfe869bd : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d376491574c94a378be979811d2e12d4 : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/217058bbc7be4f17b2612dcd1af14092 : SUCCESS in 9m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8aebe5eceb91477faac00eb8566e33b8 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/786b0ae1eeaa442589fe286aad5ec936 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b83e6386bf7e41d3bf4120afe5d54f12 : SUCCESS in 6m 20s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/dfbd81aa12bc452cb7225f3624e8f410 : SUCCESS in 42m 17s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8b052691c8604f0b85a4b3beaa0e33c9 : SUCCESS in 39m 55s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3449e2c5e48649ee8fec460e6aecabbf : SUCCESS in 43m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/23f3c691415d40d68d7302aecb415d3e : SUCCESS in 44m 53s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/ad6f0fa343054e59827a790e1d2cd340 : SUCCESS in 45m 49s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/f2087c36351d4c6cbdcb1e65eede8776 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b71fca8fb28a292d02ae572c4c62af12459ae848","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-04-19 07:23:14.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3979e452c41d5e14246f42e8baba7336c3ecffe5","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2022-04-21 11:05:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5536b67efbae624f1339959cdd4ea313438a2dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:06:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0977a1961ab89e14e319bd8ff4d6bc1f7baab858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 12:02:41.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"fb04f7f536ff9978abb5ec0f9fa4e7ee43c1a073","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-06 02:37:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acaaede5c7830f81277864adbfcc67272940f173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 03:49:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c582e5030e24f88aa6e66dc72448e65 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9032cf664d25490a972a8cc67fcb099e : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3432430d0914d7182124247bdeee7ed : SUCCESS in 9m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/178c69d4750c406ab0dd3c3a39df602a : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/822b6f6165fd4b068f0021c818b2fa45 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4912f40035fe4903b383853de99acaf7 : SUCCESS in 5m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/1b99801323b4470f96df40ad32338154 : SUCCESS in 52m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/012e0cd185ce4c8499e6458eacd114d7 : SUCCESS in 37m 17s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/db9770f54df94ff3ac5520c5b51a4f06 : SUCCESS in 57m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8999aeb2cb09436495463ff68e331475 : SUCCESS in 47m 52s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/dce71c6fe4d74eda87505836afcdb212 : FAILURE in 16m 29s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/3e36554fdd754091873627db11fc7d28 : FAILURE in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e163fae21aeeb0b5986c6e49329354e2e08ec854","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-13 05:58:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ff0caf7f036d174ebd007807fa528fa1647372b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 06:59:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02356ba2f3cd4a97aaccb2ef3d435e51 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9376656a47244784a4a9d7c072e4b0ad : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5e2a86b4e31410fb0f3ef870c9a0ed4 : SUCCESS in 8m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ae7bc45eb6547858b7a2ec46fefeb61 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3aab7de1af54d8baa8dead3b9dcd67d : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc40307e3f694d56bdaa518ed54389e2 : SUCCESS in 5m 20s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0bd6d81ef27e4e80bbab0d46639e5076 : SUCCESS in 36m 45s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/bb0b5fb9374f46ffb1ea420076f033e3 : SUCCESS in 45m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ae372ff73b894fbb8708d116820e1680 : SUCCESS in 40m 42s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/7a542f7566454f5dadc6d3084568dca6 : SUCCESS in 42m 34s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/50d250d8309847b99c1b88c210e95f6c : SUCCESS in 55m 46s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c120615e1c0347a49744ddfdd0cd5f6b : SUCCESS in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fef9d2570f22a2c7183383717e1e576ca65f0241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 06:59:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ec4392c27f0b43178592c7f4851dc0ddee67c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 07:55:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0156080d68d8431dbf4447c3115c3300 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3556d74fc54f4510a87c72cc2f57d92d : SUCCESS in 8m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7c06e82b29e48fca9fe3433c7ef931f : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96348e9135046baaeb1f04a1280db32 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c574fa74a91454eafd39ada2bf995da : SUCCESS in 5m 07s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e5accce6b6a241fd90de6c383707dd76 : SUCCESS in 47m 43s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3bc00b986f55419dbeecb6cf55adc1f8 : SUCCESS in 45m 10s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e242d241b9ff4b83995b5f50847f6f04 : SUCCESS in 44m 47s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1cc851b5545f41178b10949e78c260e4 : SUCCESS in 46m 31s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6ebbc378803747bf911757de39fceb89 : SUCCESS in 55m 53s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/bbe6cc151ead476393479202dcad7e34 : SUCCESS in 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd962b631faaf93e665797bb8960b29df1768c0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 07:55:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ea971a601a293341db33e4b78429c0bc9eeba555","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 07:57:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8de71726b164961b18d0c681849e7e9 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/129d9706aa024bf180539374e3e2244c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ff2b4e8f4a54e4024d55499f49c1f4e67b57b0e","revisions":{"9ff2b4e8f4a54e4024d55499f49c1f4e67b57b0e":{"kind":"REWORK","_number":1,"created":"2022-04-14 12:51:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/837864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/64/837864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/837864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/837864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/837864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/64/837864/1"}}},"commit":{"parents":[{"commit":"12ee78cdef1b3ca26ada16f05af795f5ad4f3da4","subject":"Deprecate unused [DEFAULT] http_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/12ee78cdef1b3ca26ada16f05af795f5ad4f3da4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-14 12:48:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-14 12:51:03.000000000","tz":540},"subject":"Migrate evaluation_interval to [evaluator]","message":"Migrate evaluation_interval to [evaluator]\n\n... because this parameter is specific to aodh-evaluator and not used\nby the other services.\n\nChange-Id: I6a78a749c2b2f03195dae3390c54663632cc8069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9ff2b4e8f4a54e4024d55499f49c1f4e67b57b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9ff2b4e8f4a54e4024d55499f49c1f4e67b57b0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
