)]}'
{"id":"openstack%2Faodh~841001","triplet_id":"openstack%2Faodh~master~If91d94e9cce4ab8e060b334f6e7f033f4ef66df9","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-08-22 07:24:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If91d94e9cce4ab8e060b334f6e7f033f4ef66df9","subject":"Update python testing as per zed cycle teting runtime","status":"MERGED","created":"2022-05-07 04:53:22.000000000","updated":"2022-08-22 07:26:11.000000000","submitted":"2022-08-22 07:24:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841001","meta_rev_id":"6fe32fdc2df364cfdbc1c40ccdfa82990d9f5f69","_number":841001,"virtual_id_number":841001,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 07:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-08-22 07:10:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-08-22 07:10:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-07 05:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 07:10:45.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"bdeb6673107e9d7d4137915b3e4a4003b2af3a07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-05-07 04:53:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bde89ba5841b08e4f4a20934a5f18488e48c5e69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-05-07 04:55:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7dce5960cc0a7e034d19dab445212340a21143a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 05:05:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/791b09359cce428fa605711dd7f216a1 : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c4ea8ed9654530b31732a2be23a74a : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72ab1e35dc344d028da5ec3d4d9d63bf : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8edbc27316c94da5a998c7e818da26ac : SUCCESS in 7m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90435e1d7818491f98b5cde3357901c8 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3316db30df8d4c0b8639ca1702c20122 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a486b1748b4772b30e7aea1aec69d107cca6d521","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-08-22 07:10:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"04259dc6366e10c0465b581649ceb46132f24bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 07:11:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e1f07551838f7c9ed49b6f35b5f61c8b4a13788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 07:24:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6fb1a78462c4af7b339a89ae705e95c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/518c8f736bb14a348e1ac7a3e611ff0a : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06f04a2ab8542d786784c0c8a13b698 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83d5d0e4cbb14ba2941542becf1a5022 : SUCCESS in 8m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efcaa28e9a0144de949fe400f4118351 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5364b34f754143628d351a8419c10e72 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":2},{"id":"995862874a6c6000bbf892c9f341074639cf5927","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 07:24:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6fe32fdc2df364cfdbc1c40ccdfa82990d9f5f69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 07:26:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d43dfeb422147c48483b6f819fa84db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c1b11b83f44cd9ba96c0b08ca2d962 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8064a871d2f923bcf317619999ee7ddaed5f9db2","revisions":{"387ae6a0389dda270fdea82e3e07ddbcfc7d5c09":{"kind":"REWORK","_number":1,"created":"2022-05-07 04:53:22.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/01/841001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/01/841001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/841001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/841001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/841001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/01/841001/1"}}},"commit":{"parents":[{"commit":"ddf12d559d012015d4b3d35dc0f7e4aac4660807","subject":"Remove unnecessary unicode prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ddf12d559d012015d4b3d35dc0f7e4aac4660807"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-05-07 04:52:11.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-05-07 04:52:11.000000000","tz":480},"subject":"Update python testing as per zed cycle testing runtime","message":"Update python testing as per zed cycle testing runtime\n\nChange-Id: If91d94e9cce4ab8e060b334f6e7f033f4ef66df9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/387ae6a0389dda270fdea82e3e07ddbcfc7d5c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/387ae6a0389dda270fdea82e3e07ddbcfc7d5c09"}]},"branch":"refs/heads/master"},"8064a871d2f923bcf317619999ee7ddaed5f9db2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-07 04:55:10.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/01/841001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/01/841001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/841001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/841001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/841001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/01/841001/2"}}},"commit":{"parents":[{"commit":"ddf12d559d012015d4b3d35dc0f7e4aac4660807","subject":"Remove unnecessary unicode prefixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ddf12d559d012015d4b3d35dc0f7e4aac4660807"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-05-07 04:52:11.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-05-07 04:55:10.000000000","tz":0},"subject":"Update python testing as per zed cycle teting runtime","message":"Update python testing as per zed cycle teting runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Removing the py36 centos8 job as well as\nupdating the python classifier also to reflect the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: If91d94e9cce4ab8e060b334f6e7f033f4ef66df9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8064a871d2f923bcf317619999ee7ddaed5f9db2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8064a871d2f923bcf317619999ee7ddaed5f9db2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
