)]}'
{"id":"openstack%2Faodh~845197","triplet_id":"openstack%2Faodh~stable%2Fyoga~Ief75ccdc0fe5c6a197fc05bc481a5be096447558","project":"openstack/aodh","branch":"stable/yoga","topic":"bug/1974682-stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-28 09:36:39.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-09-08 11:58:18.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-09-28 09:36:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief75ccdc0fe5c6a197fc05bc481a5be096447558","subject":"Ignore Gnocchi API error when the metric is not yet created","status":"MERGED","created":"2022-06-09 13:12:26.000000000","updated":"2022-09-28 09:37:42.000000000","submitted":"2022-09-28 09:36:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"845197-bug/1974682-stable/yoga","cherry_pick_of_change":843764,"cherry_pick_of_patch_set":5,"meta_rev_id":"029a5c737abae68e82935af95170b223b4e439eb","_number":845197,"virtual_id_number":845197,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-28 09:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-09-08 11:58:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-09-08 11:58:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-09 14:05:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-10 04:47:59.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"CC"},{"updated":"2022-06-10 11:48:35.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2022-06-23 23:51:55.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"state":"CC"},{"updated":"2022-08-16 19:16:15.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"CC"},{"updated":"2022-08-18 22:02:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2022-09-08 11:58:18.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"d934d9b7977c7895a37e59fc6fe30d16498df812","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-09 13:12:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8a429d8d970523d480c76c7f22d08e06d5358033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 14:05:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e4af76ee604e5aacb67251c3ed583d : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74d56cc45f7e45d4a22f72326d3602dc : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ad5a134e0584bc79f28b580982b2537 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4d3173009a4e45afa6a9c49fdd1f2b : SUCCESS in 8m 28s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/589f24d65c1c4f47936c93fa23c6311c : SUCCESS in 36m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f725a7392b1e4f9f90d0f54408acf7a7 : SUCCESS in 36m 02s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c5282d27f4a44b6fac34250357a797b4 : SUCCESS in 44m 14s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3ae38e089c2c487ea8cb6ac6f4d08892 : SUCCESS in 45m 05s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/c74afdf6fda94b22a2a9fa09bb61ffec : FAILURE in 47m 49s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/7213807916ff4f2693cc91f3b349ce36 : FAILURE in 37m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f8d2cd39c8e5e8181ea448471aa92e923801ddc1","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2022-06-10 04:47:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1365dfd2c106818e960cb61b723feaf75ea8e6d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 05:32:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b859d403ccff45fbb746cfe8cc00e5e3 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf10d4d2ecc42a49aaa8a9d2162a502 : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b726183d5575485d828789d12ed47919 : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063cbe8611444270ab87bd82d0f0d02b : SUCCESS in 9m 36s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/7ce77c9fc019476dba944d8d98d17e45 : SUCCESS in 36m 56s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/93f9f4e1141443de817824974c7351ea : SUCCESS in 35m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a5a95480c9ca49ff8ceab847a661a582 : SUCCESS in 33m 35s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/956b8c1201664799aa9524b169883919 : SUCCESS in 41m 53s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/2446497195324f55979bff8afc6f39c0 : FAILURE in 32m 32s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/c84db87b2b044ca99d137d0e44ebdc58 : FAILURE in 41m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b34be955bc290b8c27083c2246117b2e66466c24","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-10 11:48:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10f7a46b84910a65b2c2cb80d92404a217066ba3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-10 13:01:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ec7ec839dacf505b8decf22699e6f2f4d1affe0a","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2022-06-23 23:51:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4827cb841c4d0448ba05d813050c5288f8b9e5b","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-16 19:16:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d0bd49c357631d2a4d06d89433765522f430b69","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-16 19:19:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"260c3c97ac7d722bf761cdb1092dfbeef656f958","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-18 22:02:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca31b259f000296d94388931298f2a6c28c518af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-09-08 09:35:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"185f4e53b312b7a0ea06e1bd62fd6972cf67dafd","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-09-08 09:37:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e033a911c8c024ac1bf909f4bd2a3c717061684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 09:54:11.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/853829 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"5a19d06c999b76330a7e690b5bdfb5ad8f8d43ba","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-09-08 10:18:38.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1311cbf60e3170e4ad9509dc1b534de7dee23941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:21:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e4ecd95b7c94da9bc51cfabb25365b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63442c67272644809caeae294d1a62d7 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18e37a1e177d4a5286e11f773c898a88 : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77c9dbe34f334963a6eb1b8cddc459f5 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa937f200044d61aae8ae371cd24ba5 : SUCCESS in 9m 20s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6d2c907abcb44c7ba033cc40f15ce282 : SUCCESS in 45m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8e73be6645ea47108e5acca3b5949cc2 : SUCCESS in 44m 00s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cfb9a9cf1e714099971143aab1d5ebc1 : SUCCESS in 47m 58s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/aeb22e9cb0314977bb6b1036ac921222 : SUCCESS in 50m 00s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/4399d4303654422e84c4ccbeb3309b52 : SUCCESS in 54m 15s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/58790d57ccf645ac96492aebee22358f : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":2},{"id":"46325ff6d8e90a34a902b066b33e3bbb151dbe99","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-09-08 11:58:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"86b7915a6f8e3438b5a3cd9077880e36d67ade59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:58:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5fb662859d34de4b603ad01a60c0290\n\n\nWarning:\n  Change 854807 in project openstack/devstack does not share a change queue with 845197 in project openstack/aodh","accounts_in_message":[],"_revision_number":2},{"id":"d4ef9a32fbaadeb720ee2c72b0629b32662db61b","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-09-08 14:28:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ce0f219a725b809ff07150f4bdb1dcce62fd543","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-09-28 07:25:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d1df9995e87c5bdd5f7bdacd6c6df078c0f8798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 08:26:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74014d7ac03c4d5c827102c4628d9a5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32669faae6ca4b34b6a3121a267f9e97 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91afcf2f5d30479fa550d0d9d07ecb8b : SUCCESS in 10m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/397e1b45c0d741e393966faccd71ddbc : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cfb06a9e804cf5a356fef1651db3d9 : SUCCESS in 6m 55s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/97549dcefd3649b6888a320cde2fc381 : SUCCESS in 52m 39s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ed10cc8fee4949d69722c2e20b2bc4c0 : SUCCESS in 44m 47s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1847d01bf1914bfd860e23583daaea63 : SUCCESS in 40m 52s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/5c036b89effe470ea9e8fea4ffec041f : SUCCESS in 45m 48s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/08627f77382c4dbf900424cb1b4012d3 : SUCCESS in 53m 51s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/9cc74fbda6f749bdb78418587d142fec : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":2},{"id":"7ebdd6f03e1962542131df6b624ad3724d4ee607","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 08:26:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bb1392436b8cb3182233507309e45af4b57083d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 09:36:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05f5cb943e0d4859b2dc9bd1e573cc83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d8b886778b49379b5d0b51c16769fb : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4279bab016214f3a9c32695b952b61ab : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e7595110c57427b90be65336a833bec : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceee18bcfba44df58eb43ac2aca7a508 : SUCCESS in 8m 15s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2eabadf0875643d2a22f6e0c04b284c1 : SUCCESS in 48m 22s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4c652fdc1a5a4a2bab7cbc969187cdd1 : SUCCESS in 35m 13s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6716f78417cf47949af69a79f5c68154 : SUCCESS in 57m 18s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/77a4e45ff3db4a07bcffbe008acc141c : SUCCESS in 37m 34s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/72116e6eeb114d9e93e07f7d20e4253d : SUCCESS in 1h 05m 46s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/f2cbbc6193e948c3809952c81eafbc7b : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":2},{"id":"9cf6713fb6b062c87800c1f3ef7da7fe70c4a81e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 09:36:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"029a5c737abae68e82935af95170b223b4e439eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 09:37:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d1f2993b8b4ec4b91b207d15eaca3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9029d87970c4a128e23f76446d79360 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ad64b04e858565fb830ba207841b70edf5ff347","revisions":{"8e2c23f3dd77581e3dc02de40d5197e1f43048c4":{"kind":"REWORK","_number":1,"created":"2022-06-09 13:12:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/845197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/97/845197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/845197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/845197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/845197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/97/845197/1"}}},"commit":{"parents":[{"commit":"539145cce1cbb1a862f135518f8b316a3fd0002a","subject":"Add Python 3.9 to supported runtimes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/539145cce1cbb1a862f135518f8b316a3fd0002a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-30 04:33:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-09 13:12:26.000000000","tz":0},"subject":"Ignore Gnocchi API error when the metric is not yet created","message":"Ignore Gnocchi API error when the metric is not yet created\n\nChange 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8 replaced usage of\nmetric aggregate API by Dynamic aggregate API.\n\nHowever the Dynamic aggregate API returns a different response code\n(not 404, but 400) in Gnocchi \u003c\u003d 4.4 and this is causing the direct API\nerror when the metric associated with the aggregation does not exist.\n\nThis change adds 400 to the ignored response codes so that Aodh api\ndoes not return error even when the associated metric is not yet\ncreated in Gnocchi.\n\nCloses-Bug: #1974682\nChange-Id: Ief75ccdc0fe5c6a197fc05bc481a5be096447558\n(cherry picked from commit 214be56b3865d4d5bc30eaf0cb4b56bde9577f8c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8e2c23f3dd77581e3dc02de40d5197e1f43048c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8e2c23f3dd77581e3dc02de40d5197e1f43048c4"}]},"branch":"refs/heads/stable/yoga"},"2ad64b04e858565fb830ba207841b70edf5ff347":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-08 09:35:12.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/97/845197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/97/845197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/845197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/845197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/845197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/97/845197/2"}}},"commit":{"parents":[{"commit":"539145cce1cbb1a862f135518f8b316a3fd0002a","subject":"Add Python 3.9 to supported runtimes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/539145cce1cbb1a862f135518f8b316a3fd0002a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-30 04:33:06.000000000","tz":540},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-09-08 09:35:12.000000000","tz":0},"subject":"Ignore Gnocchi API error when the metric is not yet created","message":"Ignore Gnocchi API error when the metric is not yet created\n\nChange 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8 replaced usage of\nmetric aggregate API by Dynamic aggregate API.\n\nHowever the Dynamic aggregate API returns a different response code\n(not 404, but 400) in Gnocchi \u003c\u003d 4.4 and this is causing the direct API\nerror when the metric associated with the aggregation does not exist.\n\nThis change adds 400 to the ignored response codes so that Aodh api\ndoes not return error even when the associated metric is not yet\ncreated in Gnocchi.\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/853829\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/854807\n\nCloses-Bug: #1974682\nChange-Id: Ief75ccdc0fe5c6a197fc05bc481a5be096447558\n(cherry picked from commit 214be56b3865d4d5bc30eaf0cb4b56bde9577f8c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2ad64b04e858565fb830ba207841b70edf5ff347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2ad64b04e858565fb830ba207841b70edf5ff347"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
