)]}'
{"id":"openstack%2Faodh~845519","triplet_id":"openstack%2Faodh~master~I66f1d32f441c875350625d82ebf79968cacfbd9a","project":"openstack/aodh","branch":"master","topic":"telemetry-queue","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-23 06:40:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66f1d32f441c875350625d82ebf79968cacfbd9a","subject":"zuul: Use telemetry queue","status":"MERGED","created":"2022-06-13 00:57:29.000000000","updated":"2022-06-23 06:42:51.000000000","submitted":"2022-06-23 06:40:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845519-telemetry-queue","meta_rev_id":"e299c901f38ef4d64748f213c78b2248b1e10f5a","_number":845519,"virtual_id_number":845519,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-23 06:40:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-06-13 05:56:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-06-23 05:51:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-13 02:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-13 05:56:45.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"9496aa5eae8faa9128ce2f1ee46521c87598b006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-13 00:57:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03403ffbbbfd16bd43650c2f230cac11cae7a6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 02:04:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401608bc85314eb0a078fe0f8521ad32 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df3c9ee84e5d4fd8902598ef43f5e65f : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/271fb0df59414b09ac46eed71d67f025 : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c8dd74a010f4942b54b3ed95bf7f712 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae84a9b33204f3089ddde7feaa04eab : SUCCESS in 7m 14s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/60dd606f40204a1a838a352ad6016ec7 : SUCCESS in 30m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/60dac320bf5e43b6a458cc54ee99ba0f : SUCCESS in 29m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4318f66c274d4a58b8840015b8d6e8b6 : SUCCESS in 31m 09s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c3c946ce862d4fe0a8af2f9d9e7880ea : SUCCESS in 29m 41s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/be7956fda14444769075b09998fc4d33 : SUCCESS in 35m 32s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c9f2bb8bf1664c78a800f58f1a25b42c : FAILURE in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6543cb0f34e0fd4cd190b69ba9bb5730b6aa6be2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-13 05:56:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"734a96ce1adf1211bb1a2765fb3486b2abe9c23f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-23 05:51:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2463f3c7ffb46a6d56f6f72cf05718902d8a25b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 05:52:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7801cb613f32b122d05103553e018e1f3a8e871f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 06:40:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5fe0d74970b4f76b16e433458844d91 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23f784d04ab74307875d18605b7c83bc : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3ed3b0f3e534fabb40965efdac0ce99 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f08d50c0cdd45a0b865d1f5f72cec08 : SUCCESS in 9m 52s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a1973dd96f174acea6ae3e356c094f55 : SUCCESS in 40m 24s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/cd44718a6c8245aa9a13dca02698c0c6 : SUCCESS in 41m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/70c4a14a90834fbaadba335b3b333076 : SUCCESS in 41m 48s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b09b28f182f24c3cbb060b1c1cedd733 : SUCCESS in 43m 18s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/714163b3b2cd47a39cf34f50bad09cb2 : SUCCESS in 45m 07s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/223fce67df5c4c099f21a09400915c8e : RETRY_LIMIT in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd7d0639bce6c505c18433c36cb37938f5d26d6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 06:40:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e299c901f38ef4d64748f213c78b2248b1e10f5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 06:42:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23200772d2e47feacfb351a19336311 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37c274823e80adb3aa2b212ad8133f7b3f66f7f2","revisions":{"37c274823e80adb3aa2b212ad8133f7b3f66f7f2":{"kind":"REWORK","_number":1,"created":"2022-06-13 00:57:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/845519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/19/845519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/845519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/845519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/845519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/19/845519/1"}}},"commit":{"parents":[{"commit":"0b0c0aff8b8ec61a20615d34b006593ef8010fe5","subject":"Replace CentOS Stream 8 by 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0b0c0aff8b8ec61a20615d34b006593ef8010fe5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-13 00:56:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-13 00:56:39.000000000","tz":540},"subject":"zuul: Use telemetry queue","message":"zuul: Use telemetry queue\n\nCeilometer and Aodh are usually integrated, thus executing CI jobs in\nthe same queue would make more sense than merging fixes independently.\n\nChange-Id: I66f1d32f441c875350625d82ebf79968cacfbd9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/37c274823e80adb3aa2b212ad8133f7b3f66f7f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/37c274823e80adb3aa2b212ad8133f7b3f66f7f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
