)]}'
{"id":"openstack%2Faodh~853290","triplet_id":"openstack%2Faodh~stable%2Fyoga~I62b635f1ed88a3c91a1b792361c9efe4c9aaf4c2","project":"openstack/aodh","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-09-14 10:22:57.000000000","reason":"Change was abandoned"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-08-22 07:03:59.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-08-16 16:19:29.000000000","reason":"\u003cGERRIT_ACCOUNT_13177\u003e replied on the change","reason_account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}}},"hashtags":[],"change_id":"I62b635f1ed88a3c91a1b792361c9efe4c9aaf4c2","subject":"Migrate CentOS Stream 8 job to CentOS Stream 9","status":"ABANDONED","created":"2022-08-16 13:06:14.000000000","updated":"2022-09-14 10:22:57.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"bd33bb89e4903f53dce32ed8b6da6be9d4fe04c6","_number":853290,"virtual_id_number":853290,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-17 12:07:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":-2,"date":"2022-08-22 07:03:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-16 13:33:29.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"state":"REVIEWER"},{"updated":"2022-08-16 13:33:29.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2022-08-16 13:33:29.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2022-08-16 14:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2be200cc208fc82e5f20a167c125cf206a66b2f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-08-16 13:06:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c2bfdb7870ead98b2c3e5b9b18139b3d6017bf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-08-16 13:11:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e0803e7dd3555f1873d18f1a5340d7a8b2ba9c33","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2022-08-16 13:33:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ac415d3da18dd3f6376f4d7450e646cc58658cc0","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-16 14:22:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d31534743020d4a06241fd0e35fe444631cd1b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 14:31:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a86a040e0fe433f853c9f996aa56434\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed9517544fcf4dff94b82fc64a81caeb : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19a04b9eaf914069b33f20488394544b : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d8e711810544272a8a13df47f53b87f : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cde35f9125d4b34abccabfbc8374aa7 : SUCCESS in 8m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/579206f2f92e4b50b96cfbfa3b572670 : SUCCESS in 43m 55s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b96efcf6122e4df5a3b8cd7a99d2c569 : SUCCESS in 42m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3dfda5e7f2db4f92aa5da13141691d79 : SUCCESS in 50m 52s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c8922687b34d4cabaa6feeb39d633955 : SUCCESS in 51m 16s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a859e1d0f5204ca6bc76b42787890511 : SUCCESS in 1h 08m 09s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/96e51202c8e3428eb48a220f334325d0 : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":2},{"id":"38b7159e793400c6d816a883eaa5232d2d247c45","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2022-08-16 14:41:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68fe2a5bfc02062e4faf7d1e447fa9ab7f747d85","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-16 16:19:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"071ece4b99db99491b841ec1ef82fd0d289c3472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-08-17 11:05:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"be149d4df75c13944ab1e2da7a41aba37cb811db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 12:07:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6820be7732c44ea2b3228f1d1d56f5f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fafa95fa4dc493a83f78bbd481bf8fb : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62f4a3bd31be4989bd4a9dc0b35cfbcd : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/517952461b824de29988bdb021bc1057 : FAILURE in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ad7f25a59a428da25fadf83128b370 : SUCCESS in 7m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f32a373d8ecf47d8a04d7f554e68c838 : SUCCESS in 54m 40s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3d9163788cb640c2bc56d3f8c3d9a904 : SUCCESS in 42m 48s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4dad7e5d3b4f49fcb71b736767f34a89 : SUCCESS in 34m 51s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9b0c725fe8d743e3a83de7c6706d676b : SUCCESS in 34m 08s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/3dad842bb5cc45c284ed8fd2387fdc61 : FAILURE in 35m 03s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/7c79d7188c1c4dc7979b7baf3e3de673 : RETRY_LIMIT in 24m 20s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/8662de6c5cf04b9a9ccd142df72c76cc : SUCCESS in 54m 02s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/42d8250f385a47d380d4dbad979e0eb9 : RETRY_LIMIT in 26m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a907d0ee3341cef316a8178dd3ecc96985993e0b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-08-22 07:03:59.000000000","message":"Patch Set 3: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bd33bb89e4903f53dce32ed8b6da6be9d4fe04c6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-09-14 10:22:57.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/aodh/+/857480","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7d5114f6f93b7ed8279b97be1cdcdb8d94922a4c","revisions":{"a71081a848e66fd224f525f7ab6c2ccc98f999e7":{"kind":"REWORK","_number":1,"created":"2022-08-16 13:06:14.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/90/853290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/90/853290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/90/853290/1"}}},"commit":{"parents":[{"commit":"539145cce1cbb1a862f135518f8b316a3fd0002a","subject":"Add Python 3.9 to supported runtimes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/539145cce1cbb1a862f135518f8b316a3fd0002a"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-08-16 13:07:54.000000000","tz":0},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-08-16 13:08:25.000000000","tz":0},"subject":"Migrate CentOS Stream 8 job to CentOS Stream 9","message":"Migrate CentOS Stream 8 job to CentOS Stream 9\n\nChange-Id: I62b635f1ed88a3c91a1b792361c9efe4c9aaf4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a71081a848e66fd224f525f7ab6c2ccc98f999e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a71081a848e66fd224f525f7ab6c2ccc98f999e7"}]},"branch":"refs/heads/stable/yoga"},"34e2f622ea5868e74e43544c708690b0d5936175":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-16 13:11:23.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/90/853290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/90/853290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/90/853290/2"}}},"commit":{"parents":[{"commit":"539145cce1cbb1a862f135518f8b316a3fd0002a","subject":"Add Python 3.9 to supported runtimes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/539145cce1cbb1a862f135518f8b316a3fd0002a"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-08-16 13:07:54.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-08-16 13:11:23.000000000","tz":0},"subject":"Migrate CentOS Stream 8 job to CentOS Stream 9","message":"Migrate CentOS Stream 8 job to CentOS Stream 9\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/853291\n\nChange-Id: I62b635f1ed88a3c91a1b792361c9efe4c9aaf4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/34e2f622ea5868e74e43544c708690b0d5936175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/34e2f622ea5868e74e43544c708690b0d5936175"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"},"7d5114f6f93b7ed8279b97be1cdcdb8d94922a4c":{"kind":"REWORK","_number":3,"created":"2022-08-17 11:05:26.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/90/853290/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/90/853290/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/853290/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/90/853290/3"}}},"commit":{"parents":[{"commit":"539145cce1cbb1a862f135518f8b316a3fd0002a","subject":"Add Python 3.9 to supported runtimes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/539145cce1cbb1a862f135518f8b316a3fd0002a"}]}],"author":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-08-16 13:07:54.000000000","tz":0},"committer":{"name":"afariasa","email":"afariasa@redhat.com","date":"2022-08-17 11:07:27.000000000","tz":0},"subject":"Migrate CentOS Stream 8 job to CentOS Stream 9","message":"Migrate CentOS Stream 8 job to CentOS Stream 9\n\nChange-Id: I62b635f1ed88a3c91a1b792361c9efe4c9aaf4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7d5114f6f93b7ed8279b97be1cdcdb8d94922a4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7d5114f6f93b7ed8279b97be1cdcdb8d94922a4c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
