)]}'
{"id":"openstack%2Faodh~853378","triplet_id":"openstack%2Faodh~stable%2Fyoga~Ia8d5ad4246d1f823296cbbe23e1d3e55dac8c947","project":"openstack/aodh","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-15 03:17:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia8d5ad4246d1f823296cbbe23e1d3e55dac8c947","subject":"DNM: testing","status":"ABANDONED","created":"2022-08-17 00:18:42.000000000","updated":"2023-03-15 03:17:12.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e03b1df52634060ab7ed768a02df38ea0939236","_number":853378,"virtual_id_number":853378,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-17 01:14:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-17 01:14:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e5b8d8325afe955202b1f6bd29a8bd4feb08d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 00:18:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efb563bd92d17d588dd0166b8b14433786eea8a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 01:14:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af246d57acc9463da13144826793f54b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0045c1551f8d4ed8ac04e53d7416f132 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2551371674d54c0dbcd87c5d48c15a6b : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf15d6a5c064b688ac5e3b0623aadb2 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56035a7ee87c4f61a1ba47fd8a9307e6 : SUCCESS in 8m 32s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/98637a6e8c184178b6e7498daf383e4e : SUCCESS in 46m 54s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/6e0196b6dbc54cfb939dc0683b4d0e00 : SUCCESS in 42m 21s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6e931f6657464a08b319e5061ec4d6c3 : SUCCESS in 47m 30s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4a0fb8b2ba584d13bb477f572b1b65fd : SUCCESS in 47m 28s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/7da09752d878454ea17bb735379dce38 : FAILURE in 53m 12s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/b96f35e8393c42c1a8014daaf3ffda2b : RETRY_LIMIT in 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bcc21f8e306a2e87f6f69a97964856bdb84ee286","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 01:19:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3678bdbd03b0ef3dd2e7f575027acdd7a86c430","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 01:19:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b560adafc675b1626566935d713e5e05f581b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 01:21:30.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"076ee44c72585e095bb112bc73543fa43e081303","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 01:33:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04cdb8620afc7d045086ec5dc2ede03e47a52707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 02:36:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16395303fcc843e9908fd211184d42b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2556bbbe5146a995613888b6bd979f : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02fccff445c445baa6efee848ef39bca : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91ca444d11784f1fa0a7ea5e42429d44 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e40d00107040168b9d05a5d0b720e0 : SUCCESS in 7m 48s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/289722439b9943f5b7bec8c8f62613d3 : SUCCESS in 38m 53s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b078c743c8ac419ab404ae0675e6cc65 : SUCCESS in 39m 49s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bd6e4dc4f41e43179167a9595581c8fa : SUCCESS in 46m 50s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6d6a4aec379b4cacaa26a1a9e2a80d86 : SUCCESS in 35m 28s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/d2773fb000524c4e8e5387d40904d2ab : FAILURE in 38m 04s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/1620be9611fe46348b3185c5d137804b : RETRY_LIMIT in 16m 57s","accounts_in_message":[],"_revision_number":1},{"id":"bd17041ef57e89c56a5497c85077f2bc0fa4dc31","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 02:46:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4778df2f44b63c90f378a188a424bf8dce612044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 03:58:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/969889467e4c472d8fe4bcd92147e170\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a3efd695c54241b6432e9d9259d4ee : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f0129eeca6c49c7963f68f1acec1914 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1d2a812229c4817ae44e7a59a08f09f : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed170a9eee284239864af4fe112c0bd1 : SUCCESS in 7m 56s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/caec629f347f4eb98d26addbcd363c5b : SUCCESS in 47m 57s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/9e6d316412cb4da29e2f419e622dce29 : SUCCESS in 47m 18s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/52ba59ce0aee426c90a4e5704125fb63 : SUCCESS in 52m 35s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/56a02f4d06b44b7ca1a62bf0afeb576d : SUCCESS in 53m 23s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/d91c3510a77948f3b7c84b78dd63b908 : FAILURE in 57m 39s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/dc6bb543259d477eabc62eef5d19b0d7 : RETRY_LIMIT in 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"13a5a4acffc6907d073f47a54e6011f2c964e4fa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 04:21:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f824ec91845bca8b1cf2bb657faf9ef6cef15826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 05:18:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93aa994865114c409710421c2f9ed95a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/522ec2d1331d4e36a832b17d7d2b6e00 : POST_FAILURE in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db046eadc8e74c05b906c4a759106841 : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77155b77e1f1403b8d298b61eab09a1e : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a18f2f71124096afeaf5b301034f6d : SUCCESS in 8m 56s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9618c2c8f9b349c89b7376cd981df6b3 : SUCCESS in 42m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8433261aa2af4ae58d346bdf8ee877cc : SUCCESS in 44m 29s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/5da99dfcfa734d578e7d7c7af35ab101 : SUCCESS in 48m 23s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1b8d37296197431c993272b36ef86cf4 : SUCCESS in 51m 04s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/08e56b6c513e47da87f13c75fb1a543b : FAILURE in 43m 28s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/82357ee2a2b04517953e51e0fb97073c : RETRY_LIMIT in 15m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ba7d6783ff09e5c103544ef95f982df1386f1157","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-18 04:49:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a134ce5a631f85316704c9a114c5665d677d313e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 06:01:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf6c2c5953004c9e9053c064355df323\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a9639d041c446ebe1f7981d9af9946 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62a824b7ef674617b3b7baf04dd5365c : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c94ce512619a43b5b7b6cb92ba533a4b : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51050a86109a4d098ad9bbc2c52992c3 : SUCCESS in 7m 39s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2ceb71d330ee44eea77ec7eabaa0649d : SUCCESS in 53m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0494e68b7b964f53a06a77233ff88c63 : SUCCESS in 33m 51s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6dd7d9b38cf84930b9cd028c69cf03b0 : SUCCESS in 39m 09s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/90ac2f4dc7ee4e39b2fdf4a7c3bba7c0 : SUCCESS in 39m 13s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/19687683fd36425ab9983e081aea1152 : FAILURE in 38m 00s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/9b8a98f4cf6f479da46ea4673aa8607b : RETRY_LIMIT in 18m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0e03b1df52634060ab7ed768a02df38ea0939236","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-15 03:17:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"163545f743ff46c12de8944736c168cc6d639f69","revisions":{"163545f743ff46c12de8944736c168cc6d639f69":{"kind":"REWORK","_number":1,"created":"2022-08-17 00:18:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/853378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/78/853378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/853378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/853378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/853378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/78/853378/1"}}},"commit":{"parents":[{"commit":"539145cce1cbb1a862f135518f8b316a3fd0002a","subject":"Add Python 3.9 to supported runtimes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/539145cce1cbb1a862f135518f8b316a3fd0002a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-17 00:18:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-17 00:18:08.000000000","tz":540},"subject":"DNM: testing","message":"DNM: testing\n\nDepends-on: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/853377\nChange-Id: Ia8d5ad4246d1f823296cbbe23e1d3e55dac8c947\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/163545f743ff46c12de8944736c168cc6d639f69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/163545f743ff46c12de8944736c168cc6d639f69"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
