)]}'
{"id":"openstack%2Faodh~855524","triplet_id":"openstack%2Faodh~stable%2Fyoga~Id984427ee1be486837970df4e518a9d844c8df7f","project":"openstack/aodh","branch":"stable/yoga","topic":"telemetry_test_centos_8s_integration_jobs","attention_set":{},"removed_from_attention_set":{"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-12-13 20:16:00.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13177\u003e by clicking the attention icon","reason_account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}}},"hashtags":[],"change_id":"Id984427ee1be486837970df4e518a9d844c8df7f","subject":"[dnm] test fix for centos-8s integration jobs","status":"ABANDONED","created":"2022-09-01 15:00:27.000000000","updated":"2023-01-06 18:11:07.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24898bb36fc70631097ed6e299b870ee9e5c75a7","_number":855524,"virtual_id_number":855524,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-07 19:56:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-01 16:00:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 05:38:43.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"CC"}],"messages":[{"id":"34b364a39238fc46e90e8d19f9a9630b6854d4ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-09-01 15:00:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b0d0f380a99b0b93c10534b4bd38a99d5592844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 16:00:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5d7c81a8372475dad10b931696e472e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d955fe8235b4f449cb7aacb4307ae90 : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9647562dbb244d42953d37076e443b0b : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e27e70c846834471aad3f535ed4119a0 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50900b4ee14143468dded7e81ce22f1d : SUCCESS in 8m 18s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/708a63f1aafe4c93811df0e1939e0744 : SUCCESS in 38m 37s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a346954432ba474fb48b64cfd185085d : FAILURE in 44m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/965707d7eb064b9da8e79a47b65713f8 : SUCCESS in 48m 28s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/56e7ddb916a94ae4871198f812868036 : SUCCESS in 45m 21s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/e5b8927bbb5f4389a5c210accd4d4321 : SUCCESS in 57m 16s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/cbbb0a2a678549d4afcf805bbb5a8102 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":1},{"id":"22ef0c342c6f64aacf91f028aa3f4a40f76a07ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-09-07 18:31:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"90f92b82c1edcbe9360343749998c0ea78dd3f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 19:56:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e708de475f4ba39ddbfdd00a4769db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ba1a2fbabb4526bc44a57de2f05357 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5b15d334ee24174b4a1f4b7b531ae57 : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0698821df6204cee9c05b40be0d15418 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694eb9db2d5b4caaa8efc28d22797650 : SUCCESS in 11m 06s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c478c784a706473f8bf892c914fbbef6 : SUCCESS in 58m 35s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/96b71177d80c45dfb56119c001219d7d : SUCCESS in 52m 29s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b7040d47e7f848939aeacf75f2c89860 : SUCCESS in 52m 21s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f2627849569a4422a850f79de6d798df : SUCCESS in 57m 50s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/f7f9f519475940f7923c337ddb571806 : SUCCESS in 1h 06m 15s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/03c1581f134f4900ba7a0f357030ad77 : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ce0d796f3e6c1c910929addc34c428964591d908","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-10-31 05:38:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f06c61bc2a1940c615af1e525d751a6c25167beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 06:40:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af5eed73697546789e189ef342c0ce20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfaf03873801485e9e2951d5e7154dde : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d331a1f0e0c4862ad450a29aa89a3a3 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77199a71c4f445a69964e18cd86c958d : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29292f4add6465ca68a9d5700f2d66c : SUCCESS in 7m 14s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d55c4132e0d1437299b8adb790b2cfb1 : SUCCESS in 50m 42s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0d2f2613587b444dbbde24e04c28e6c2 : SUCCESS in 47m 54s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d7c461993de644599f664f7245101ccb : SUCCESS in 52m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/cfbf0999a20b4bbbb6fe569544705b98 : SUCCESS in 56m 28s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/52ac8604440d4bce922d2811c1d1411d : SUCCESS in 56m 44s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/fd5a395ac0cc49f49d26deeafb9f1754 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":2},{"id":"24898bb36fc70631097ed6e299b870ee9e5c75a7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2023-01-06 18:11:07.000000000","message":"Abandoned\n\nNot needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0043c1205ed93d6b20ce3319d511eb0df7ad427c","revisions":{"234504b01b763617943f4461d7f7b676a924e967":{"kind":"REWORK","_number":1,"created":"2022-09-01 15:00:27.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/24/855524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/24/855524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/24/855524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/24/855524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/24/855524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/24/855524/1"}}},"commit":{"parents":[{"commit":"539145cce1cbb1a862f135518f8b316a3fd0002a","subject":"Add Python 3.9 to supported runtimes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/539145cce1cbb1a862f135518f8b316a3fd0002a"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-09-01 14:58:18.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-09-01 14:58:18.000000000","tz":60},"subject":"[dnm] test fix for centos-8s integration jobs","message":"[dnm] test fix for centos-8s integration jobs\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/853829\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/854807\nChange-Id: Id984427ee1be486837970df4e518a9d844c8df7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/234504b01b763617943f4461d7f7b676a924e967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/234504b01b763617943f4461d7f7b676a924e967"}]},"branch":"refs/heads/stable/yoga"},"0043c1205ed93d6b20ce3319d511eb0df7ad427c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-07 18:31:33.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/24/855524/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/24/855524/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/24/855524/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/24/855524/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/24/855524/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/24/855524/2"}}},"commit":{"parents":[{"commit":"8e2c23f3dd77581e3dc02de40d5197e1f43048c4","subject":"Ignore Gnocchi API error when the metric is not yet created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8e2c23f3dd77581e3dc02de40d5197e1f43048c4"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-09-01 14:58:18.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-09-07 18:31:33.000000000","tz":0},"subject":"[dnm] test fix for centos-8s integration jobs","message":"[dnm] test fix for centos-8s integration jobs\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/853829\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/854807\nChange-Id: Id984427ee1be486837970df4e518a9d844c8df7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0043c1205ed93d6b20ce3319d511eb0df7ad427c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0043c1205ed93d6b20ce3319d511eb0df7ad427c"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
