)]}'
{"id":"openstack%2Faodh~857480","triplet_id":"openstack%2Faodh~stable%2Fyoga~Ie0a34d3f2481f325962dc159c3315bd8d1d89854","project":"openstack/aodh","branch":"stable/yoga","topic":"c9s-stable/yoga","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-09-16 07:33:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0a34d3f2481f325962dc159c3315bd8d1d89854","subject":"Replace CentOS Stream 8 by 9","status":"MERGED","created":"2022-09-14 10:22:42.000000000","updated":"2022-09-16 07:35:48.000000000","submitted":"2022-09-16 07:33:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857480-c9s-stable/yoga","cherry_pick_of_change":840810,"cherry_pick_of_patch_set":5,"meta_rev_id":"67adf049798061617bb08fe66fc7e2ac83e8830d","_number":857480,"virtual_id_number":857480,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 07:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-09-16 06:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-09-16 06:13:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-14 10:22:42.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-09-14 11:34:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-15 16:41:33.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2022-09-16 06:13:46.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"162ada82d6cef60258ec8fb73ad3a6944b0d0db7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-09-14 10:22:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5bd1f2268cc92544548f50dedc521b78b9c67f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 11:34:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de7be9cd03d2496e9fe342c159e674ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe73187a29a40cda33e02cbfabfd88b : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0750d3696d3442aebe0b2b0edbe8be36 : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64c21b66e3764fa68cdbdd318a00576f : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0563f5ab34149a4b640b94b628dbb8b : SUCCESS in 7m 32s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/52cad3c491d9423787667f3200868fc5 : FAILURE in 35m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/989a60d27e634d5cb59320d6c4859a8b : SUCCESS in 39m 56s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/589aeb00dd4d4082971c9e9c8272cd51 : SUCCESS in 41m 56s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a0a064dc4e1a4432ad7f7c9e0c5e7d7f : SUCCESS in 41m 41s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/dfcfda0561314848ae2f6492c648b435 : SUCCESS in 47m 28s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/b5b54241f5fc4e05bc0291d8a8abfea0 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":1},{"id":"16e1b43153b3576598f626aa0ddc68ccf4a439ab","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-09-15 16:41:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ffc6432ccda537eaaf09d5d63a55576a8950766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 17:53:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e029b7f8d6c847d4aed245406b947f34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/787a0d85f5c246b1b442b3c747a11b2e : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c9ba3f33f2e4b0e9cfa91af4bfa84b5 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b1c34c78b244230b3d3a6aea1cb5631 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db44e3bd2d4d4cf0b77c4529eb464df0 : SUCCESS in 6m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/53c5bb4287a64844859904695c2e4fdf : SUCCESS in 52m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5cde25424f4b4a769cd69e4aaeb004c2 : SUCCESS in 46m 10s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/611f194d1d2f4f2dae208b94df7b881a : SUCCESS in 53m 48s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9a0a571ef2ce4951bc4e0f266acde776 : SUCCESS in 52m 32s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/5042c2d550cd44bb8ba78c0c80d364b1 : SUCCESS in 1h 01m 38s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/42924355c4da4795a6d6bbd958589f45 : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"dd89228df18394cc80a478ccf6647b1164085e7a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-09-16 06:13:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b8656cb7998405c61f35e4385da4dd7e0ebdbb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 06:14:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e4eb6b711a4b63e25296b418ce5973142af3c80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:33:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07bb86db49224c7e92ce0f92c4c2ad1a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f95b41e103074ddea5588321abcca245 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bc3dfc61c1e4c8a8ed3d5291061d05f : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4b04bc7a4894153a77533560d7ad782 : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26e6f39c7bf4b39935231e6398e5370 : SUCCESS in 7m 17s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f03679f1752b4816bf8ceeee4fa02926 : SUCCESS in 49m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e032c224f0234ca4b37068a995c8f430 : SUCCESS in 47m 24s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8c62acb8b9874b6d9598e382df218ae2 : SUCCESS in 58m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d105ac1d6b9d4a0c832f0b4ab239836d : SUCCESS in 52m 37s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/f7efe6ce181e49758ced3af1d63e6d9c : SUCCESS in 1h 04m 25s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/7d1668ecd3654d8dac5bf78ec34b6f10 : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"47f5a70c8314f8ca48104012994f20f5d6bc25a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:33:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"67adf049798061617bb08fe66fc7e2ac83e8830d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:35:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2c07b16ca54faeb66a4646e8010ac7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac41ca7292a42058598ed43cdbee080 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"510dc2db35df6a7fd46b1b8c284960d583097f63","revisions":{"510dc2db35df6a7fd46b1b8c284960d583097f63":{"kind":"REWORK","_number":1,"created":"2022-09-14 10:22:42.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/80/857480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/80/857480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/857480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/857480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/857480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/80/857480/1"}}},"commit":{"parents":[{"commit":"dacce7bb45dd75aa0276a88511bec713b632ff43","subject":"Merge \"Revert \"tox -e py39 hangs on stable/yoga\"\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/dacce7bb45dd75aa0276a88511bec713b632ff43"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-06 03:50:06.000000000","tz":540},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-09-14 10:22:42.000000000","tz":0},"subject":"Replace CentOS Stream 8 by 9","message":"Replace CentOS Stream 8 by 9\n\nThe latest cinder requires Python \u003e\u003d3.8 which is not available in\nCentOS Stream 8. This bumps the base OS version to 9 which has Python\n3.9.\n\nDepends-on: https://review.opendev.org/840811\nChange-Id: Ie0a34d3f2481f325962dc159c3315bd8d1d89854\n(cherry picked from commit 0b0c0aff8b8ec61a20615d34b006593ef8010fe5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/510dc2db35df6a7fd46b1b8c284960d583097f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/510dc2db35df6a7fd46b1b8c284960d583097f63"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
