)]}'
{"id":"openstack%2Faodh~860656","triplet_id":"openstack%2Faodh~stable%2Fxena~Ief75ccdc0fe5c6a197fc05bc481a5be096447558","project":"openstack/aodh","branch":"stable/xena","topic":"bug/1974682-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-10-12 08:33:56.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"hashtags":[],"change_id":"Ief75ccdc0fe5c6a197fc05bc481a5be096447558","subject":"Ignore Gnocchi API error when the metric is not yet created","status":"MERGED","created":"2022-10-07 15:10:27.000000000","updated":"2022-10-12 09:44:20.000000000","submitted":"2022-10-12 09:43:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860656-bug/1974682-stable/yoga-stable/xena","cherry_pick_of_change":845197,"cherry_pick_of_patch_set":2,"meta_rev_id":"d33a9152f3d3682a97fcb312cc7127afabd26d6c","_number":860656,"virtual_id_number":860656,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 09:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-10-12 08:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-10-12 08:33:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-07 15:10:27.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-10-07 15:56:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 07:33:50.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"CC"}],"messages":[{"id":"77c030042240fce6d90a3e47247745f7a8d704ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-07 15:10:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"0781134b6e5259096e1e33913fb523941417e99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 15:56:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e319bf4f00244063810ef59ba0d62ea3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbcefa9fc0374a08b46f399b72c2836c : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3459b948dff54376b19041fef10a4ce6 : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b80a64d41728469a913f42c3aa123222 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff6a2e37f7ac41fb9726030517c31c46 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff2ec2a45774a6f93b6d0802364aa9e : SUCCESS in 7m 54s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0e1b578e9d8e489d871c9ff1633d3cca : SUCCESS in 38m 39s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b9a01525035a4e2a866b4a52515a0b78 : SUCCESS in 37m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d585e3134783432bb488a3faad3524d8 : SUCCESS in 43m 43s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c4410808b8c6444eb5ef83a316aa3dc1 : SUCCESS in 44m 07s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/4db77cb972ef447a8b1d0143b8aa9078 : FAILURE in 9m 01s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/8e78317e9216402cadbcf425faec8de3 : FAILURE in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b11ee381e87d3bf154b5ed67b1e6fffe720fa21a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-10 06:43:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8001ad95a2eb59beef1a05929838efa5dbdfff9c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-10 12:59:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"309a10c6c49b247dd74c84cbc8533771b170e7f8","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-10-12 07:33:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41a6c003a39ca59ddec83627031a362d6749fafa","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-10-12 07:34:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e9f5f23aded6d34fc16548347e3adab7af0321d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 08:25:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3071daff174de2ba1b37f8b346b48a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6aa6c23d3ad41ed8efffcbe38c8a758 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c683f1cc1fc94b21be6553d614969305 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e2c41dce94d46aab1c96922ec7c4731 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9741cecc7044c898b4654c377966c0b : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/552fe533bc344f778d045d39d106b715 : SUCCESS in 5m 40s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/dba21d1f45ba4ed29e3fbd6ddbb9d841 : SUCCESS in 25m 55s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/67f1d9bc277149a2873bbc1c1067ece3 : SUCCESS in 29m 08s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/94e71442869346ab9bdb651a266abe21 : SUCCESS in 27m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a5fa2aee72ca4cfd91428a5baf3ee5c0 : SUCCESS in 27m 44s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/98723c31c1aa4ca688a1ceab5ae63ae0 : SUCCESS in 45m 45s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/ee21e86a88d743f6b9ccae6c3421b37e : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8583970a755bfd41b2a42721d166a08225fc2d3c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-10-12 08:33:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e00846613dd33d0721dbe181b4d3ec2a21768ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 08:34:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"536456eba2b4d1bfb56ee05c4f52ff8585cc220d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 09:43:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b60556d974cc438f8f6b5b9eb47f0c6c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e79fc42dc3a141ce82acbd5ba1e6426c : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98aceb4c9cc4420ea5efe5548e08e068 : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c552ebd0f3b746daa91fb31cffcca06f : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdef6ca9c0644933a35e6b6b8cc78710 : SUCCESS in 7m 02s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ff09f5679b504b9d9c0ecd0bee9f978e : SUCCESS in 47m 08s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d18bab3354ba4c7e873dcec9dec1bdba : SUCCESS in 48m 45s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/56b2e1089eea4a8589d28dee09cdd282 : SUCCESS in 54m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/dd77a52666c74696abc63631faf6a5d3 : SUCCESS in 53m 28s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/c7b39cb32104418c80f2399c70187984 : SUCCESS in 1h 01m 17s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/2b7575d4723940d19c7708c1c1dd43a5 : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c15f3f4449b981757e8638c36d4ca9808ff07f1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 09:43:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d33a9152f3d3682a97fcb312cc7127afabd26d6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 09:44:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc0550d0b81a485c89bb20af99324f8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c2af6d55624c6fbd1eac41d9543783 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"231f404ad13bac69964f19b9b1bfa5a42c1f977a","revisions":{"231f404ad13bac69964f19b9b1bfa5a42c1f977a":{"kind":"REWORK","_number":1,"created":"2022-10-07 15:10:27.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/56/860656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/56/860656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/56/860656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/56/860656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/56/860656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/56/860656/1"}}},"commit":{"parents":[{"commit":"71ce784c423e699332b6a7b902a26ad186432c06","subject":"Bump minimum version of gnocchiclient for aggregats API support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/71ce784c423e699332b6a7b902a26ad186432c06"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-30 04:33:06.000000000","tz":540},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2022-10-07 15:10:27.000000000","tz":0},"subject":"Ignore Gnocchi API error when the metric is not yet created","message":"Ignore Gnocchi API error when the metric is not yet created\n\nChange 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8 replaced usage of\nmetric aggregate API by Dynamic aggregate API.\n\nHowever the Dynamic aggregate API returns a different response code\n(not 404, but 400) in Gnocchi \u003c\u003d 4.4 and this is causing the direct API\nerror when the metric associated with the aggregation does not exist.\n\nThis change adds 400 to the ignored response codes so that Aodh api\ndoes not return error even when the associated metric is not yet\ncreated in Gnocchi.\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/853829\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/854807\n\nCloses-Bug: #1974682\nChange-Id: Ief75ccdc0fe5c6a197fc05bc481a5be096447558\n(cherry picked from commit 214be56b3865d4d5bc30eaf0cb4b56bde9577f8c)\n(cherry picked from commit 2ad64b04e858565fb830ba207841b70edf5ff347)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/231f404ad13bac69964f19b9b1bfa5a42c1f977a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/231f404ad13bac69964f19b9b1bfa5a42c1f977a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
