)]}'
{"id":"openstack%2Faodh~863798","triplet_id":"openstack%2Faodh~stable%2Fwallaby~I4f631d224404460138f4050b1b981d577b592544","project":"openstack/aodh","branch":"stable/wallaby","topic":"bug/1946793-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2022-11-10 09:32:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f631d224404460138f4050b1b981d577b592544","subject":"gnocchi: Use Dynamic Aggregates API","status":"MERGED","created":"2022-11-08 10:26:17.000000000","updated":"2022-11-10 09:33:52.000000000","submitted":"2022-11-10 09:32:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863798-bug/1946793-stable/xena-stable/wallaby","cherry_pick_of_change":837053,"cherry_pick_of_patch_set":3,"meta_rev_id":"3e9d4f038d45f05b09524106968fad803f205587","_number":863798,"virtual_id_number":863798,"owner":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-10 09:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-11-10 08:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-11-10 08:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 10:26:17.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"CC"},{"updated":"2022-11-08 11:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-08 12:31:14.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"38497f33636d5b01e17b320664fe527e8ea8c7a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-08 10:26:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"a50f73a0077b8473ad918305e11f0b10d4b8290d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-08 10:26:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3a0793d4517b30f3d57eb11ad4863b92905f0df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 11:43:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a09465318794732aa2e54e2a4a5fc32\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2692a0b75df94834bd1fb4d504cfafb9 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc1d6bd2e19541b1b44b88d56ae31e71 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a10567347ba44b41ae60d27ebf87e423 : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6675862dfa947188986526ff179b46b : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f80db87682406ea871416511705ab7 : SUCCESS in 6m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/93c28b04c9b74f9c8745f98b48723f33 : SUCCESS in 37m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4d640fe1057b48469c6e2b49aeeba769 : SUCCESS in 56m 38s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ee34749a67494588b823a2afc5751987 : SUCCESS in 54m 07s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1f7715bbf92a4e7ca770a56fe96f2591 : SUCCESS in 45m 13s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/6eec99a75c5c4fa599f104e041d57478 : SUCCESS in 57m 43s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/67e85d1ccad54b5c9b031e4d7b5ed678 : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":2},{"id":"13466249dad904b0cdb2e8250cf047eb3ae2b690","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-08 12:31:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9952a44a813574c3bd083c0bc7ad32cda79f3327","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-08 12:33:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"356ad0442ac5df93181955091765be7e2b86ce11","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-10 08:35:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"931fad57ebd341badff0bc9f73d4f4c3671bd39f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 08:36:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b41096b26091b4f5e19fd73af4407864aafb04b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 09:32:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4204c522fbc1460095bfd3b2a76dabc8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/368455ff23e846b98c78635fcc8dd747 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b35bcfe528414defaa29aca4e7726182 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b33d6dfd2091476a94d36db1790b3bad : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b5f39e3d2e46ff87661bb935a1819a : SUCCESS in 6m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/65690e5627c640c59678ebf6a27bc7db : SUCCESS in 46m 09s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/78de5d299c8648cf9826239265bb2702 : SUCCESS in 44m 59s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6c1d0ec932da4eec9930ce7fbe18a699 : SUCCESS in 53m 55s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9838ad3aa03b4c9495bf1cf8afe27cae : SUCCESS in 45m 16s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/b9f5363fe94c4022919b7e087a539d66 : SUCCESS in 55m 26s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/698af999004247b787d258fbb95c2d9a : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":2},{"id":"40f3ad959be1193461367746bf555a45c08077ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 09:32:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3e9d4f038d45f05b09524106968fad803f205587","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 09:33:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a8f1c9356d48de9a9568d24afee310\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e9402785a14f099a0b993475b5f72b : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"27ee81a37acf097f572ffce1c4cd72f9987e0520","revisions":{"9be2178bfd7de2ccf6734ea2c336ba3420093b24":{"kind":"REWORK","_number":1,"created":"2022-11-08 10:26:17.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/98/863798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/863798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/863798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/863798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/863798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/863798/1"}}},"commit":{"parents":[{"commit":"865d6d9cdb796aba8c4edf08363cd73869776c74","subject":"Replace oslo_utils.fnmatch with fnmatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/865d6d9cdb796aba8c4edf08363cd73869776c74"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-08 10:26:17.000000000","tz":0},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\nDepends-On: https://review.opendev.org/863284\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n(cherry picked from commit e6d55b1ea9ee852cc56002c865c9443b526f2f58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9be2178bfd7de2ccf6734ea2c336ba3420093b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9be2178bfd7de2ccf6734ea2c336ba3420093b24"}]},"branch":"refs/heads/stable/wallaby"},"27ee81a37acf097f572ffce1c4cd72f9987e0520":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-08 10:26:26.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/98/863798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/863798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/863798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/863798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/863798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/863798/2"}}},"commit":{"parents":[{"commit":"865d6d9cdb796aba8c4edf08363cd73869776c74","subject":"Replace oslo_utils.fnmatch with fnmatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/865d6d9cdb796aba8c4edf08363cd73869776c74"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-08 10:26:26.000000000","tz":0},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n(cherry picked from commit e6d55b1ea9ee852cc56002c865c9443b526f2f58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/27ee81a37acf097f572ffce1c4cd72f9987e0520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/27ee81a37acf097f572ffce1c4cd72f9987e0520"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
