)]}'
{"id":"openstack%2Faodh~867305","triplet_id":"openstack%2Faodh~stable%2Fvictoria~I8e34872b550c3a2fe0bb251f4686b55cfe6e8733","project":"openstack/aodh","branch":"stable/victoria","topic":"fnmatch-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-12-19 06:45:43.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-12-19 07:44:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e34872b550c3a2fe0bb251f4686b55cfe6e8733","subject":"Replace oslo_utils.fnmatch with fnmatch","status":"MERGED","created":"2022-12-13 21:07:52.000000000","updated":"2022-12-19 07:45:43.000000000","submitted":"2022-12-19 07:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867305-fnmatch-stable/wallaby-stable/victoria","cherry_pick_of_change":855563,"cherry_pick_of_patch_set":1,"meta_rev_id":"ca0a2cb0cc389b6a42830da1aec333d8eecf10b7","_number":867305,"virtual_id_number":867305,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-19 07:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-12-19 06:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-12-19 06:45:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-13 21:07:52.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-12-13 21:55:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-19 06:12:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"f1a4f4fa325c38c30b65b76d5edd3144d3cd9acc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-12-13 21:07:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"0a534e581639ccd090d06831840110845d50ae84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 21:55:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a02a3f805704b2fb86974d0380c5383\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eea285a6e6c446ea7f26c22450e1226 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13a914e7fd2748eb8bc5d5d3dea836c4 : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde47428c6a144e8bf8fe5864793abbe : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262587ce054c41f0828e93bae467b90b : SUCCESS in 9m 00s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/1103358f3a2942e69cde9387e66e22ad : SUCCESS in 38m 40s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2c12c1aed54b40b7a187442d6a4dc47b : SUCCESS in 33m 08s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/67f4a5fe4ff9402593a844c12667c08e : SUCCESS in 45m 35s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6a439e63f1e64ad49c86f9b8dffd50d0 : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ba680d58087e17a8c26462354ca31f4bd12c6847","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-19 06:45:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a725acea2b9eb03ef624909289bb9c5f70944cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 06:46:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9da0afb50375c9e822c2907fc62b90acd37dead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 07:44:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/973d069f48f2430085181aa5693d28a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c9bdca0a0cf49c0918ed1ad6815bd03 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d7417407ed64635a0b4189abe1e2136 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77c4d703a1dc46a08f623134662d67c2 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1359560da69a48fb94078724447197b6 : SUCCESS in 6m 28s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9ba82a48983f499eaad87c5b2cd30d05 : SUCCESS in 51m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2f37a3f22b4a4f9e8f7cefa354250163 : SUCCESS in 51m 54s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8d2bbb60acf3433cbe4b562322283608 : SUCCESS in 34m 06s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9f0edceb08034e46a67e236a562d2fe8 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8edb9e9c44258f8c3b8dc908004ba883c3d6c5ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 07:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca0a2cb0cc389b6a42830da1aec333d8eecf10b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 07:45:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9243b8e3f5ea45dea71c4d2e3e8adb21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad81fb70c0b84fe781cb0b6097406521 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e73e6111e7173604bcb1ef28a285e4d47fbc72f","revisions":{"7e73e6111e7173604bcb1ef28a285e4d47fbc72f":{"kind":"REWORK","_number":1,"created":"2022-12-13 21:07:52.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/05/867305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/867305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/867305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/867305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/867305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/867305/1"}}},"commit":{"parents":[{"commit":"028561741e125f1b343ae5bf1021df1012e92005","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/028561741e125f1b343ae5bf1021df1012e92005"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 12:06:27.000000000","tz":540},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-12-13 21:07:52.000000000","tz":0},"subject":"Replace oslo_utils.fnmatch with fnmatch","message":"Replace oslo_utils.fnmatch with fnmatch\n\nThe oslo_utils.fnmatch module was added to solve an issue in py2.7 but\nit is no longer required because py2.7 is no longer supported.\nThe module was deprecated since oslo.utils 4.9.1[1] and the stdlib\u0027s\nfnmatch module should be used instead.\n\n[1] 4c893c92f551c9dd2a7cfbe7ae8171ad8139df0b\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/853829\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/855344\n\nChange-Id: I8e34872b550c3a2fe0bb251f4686b55cfe6e8733\n(cherry picked from commit 865d6d9cdb796aba8c4edf08363cd73869776c74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7e73e6111e7173604bcb1ef28a285e4d47fbc72f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7e73e6111e7173604bcb1ef28a285e4d47fbc72f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
