)]}'
{"id":"openstack%2Faodh~867591","triplet_id":"openstack%2Faodh~stable%2Fwallaby~I074ef0ad4578abde73ae82efb18bc9edbceb7e81","project":"openstack/aodh","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-14 07:55:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I074ef0ad4578abde73ae82efb18bc9edbceb7e81","subject":"Remove TripleO job","status":"MERGED","created":"2022-12-14 03:29:27.000000000","updated":"2022-12-14 07:57:07.000000000","submitted":"2022-12-14 07:55:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867591","meta_rev_id":"f21a15ce5fd464cd9461a1a7b626aaced00d7633","_number":867591,"virtual_id_number":867591,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-14 07:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-12-14 07:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-12-14 07:02:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-14 04:28:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-14 07:02:40.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"697889682c0c38f7491b55e06c52f77a10953bb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-14 03:29:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b41ecadf6d4230d8004b3026918de7f4c075c982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 04:28:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc2d35dd291a4d1dad86f3afe1313562\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac09e9cbdeb14277b3eaab26c1fc3889 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc80ed7f01b74cb3865764469bea2a20 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/223c3d616dab45408029539883812a47 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edffb9fecfa14126a831305f74bcb646 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d50c1603667458b985b532f96c3f84d : SUCCESS in 7m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f23090b1175541839599d741f66df380 : SUCCESS in 45m 56s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e5ea54126c8a42068739382a4889504f : SUCCESS in 45m 21s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3cf883a205c448ca91e5915c710dc8be : SUCCESS in 51m 24s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/44300c53395946a8a9fd4b5fc16aec6b : SUCCESS in 49m 54s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/8fd2af30126c458cbb9891f81cd5e32b : SUCCESS in 55m 58s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/5522c34e12114ad3934573778a4af252 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"86462af20c2705f1b6bdad15bb16964766a0ac07","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-14 07:02:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70e33310a2c449263d9bf321106183379cc1c361","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 07:03:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6eb86dc3a9e02a2858067362265a68d489bd2239","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 07:55:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee7d264f15214c089504b99c0cefad33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f312ca2cdd194e06986a22d34b624af8 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0de76ecfd4ba4d9c8549ef244134612f : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/661897648479455084e29e559a6a4c37 : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6826e398f0754266aae4284b41f1a4f8 : SUCCESS in 8m 08s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5aa0518467074f4296b2c16e2582a921 : SUCCESS in 40m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5896428834e84928903ec84d03fe5645 : SUCCESS in 35m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/34b1d5a6c2af4a53bfa2a6eb3bd0ac66 : SUCCESS in 43m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f4e9e02bc73f487ca762de764a6ad332 : SUCCESS in 43m 57s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/67eb28712db74084adf948e642924212 : SUCCESS in 51m 05s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/662942cdb7e34d8cab6190945983267c : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e73adbdd2c4e4c88a3a54f259281db0bdf9afe6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 07:55:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f21a15ce5fd464cd9461a1a7b626aaced00d7633","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 07:57:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8c4d34bafd348fe8fa69ce6d85ee094\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2cc01287574098a7e867f608a2b70f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4cae54cd9bbd1ce3cf2bec2ad8d28ffb9ea381e","revisions":{"b4cae54cd9bbd1ce3cf2bec2ad8d28ffb9ea381e":{"kind":"REWORK","_number":1,"created":"2022-12-14 03:29:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/867591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/867591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/867591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/867591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/867591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/867591/1"}}},"commit":{"parents":[{"commit":"865d6d9cdb796aba8c4edf08363cd73869776c74","subject":"Replace oslo_utils.fnmatch with fnmatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/865d6d9cdb796aba8c4edf08363cd73869776c74"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-26 02:03:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-14 03:28:37.000000000","tz":540},"subject":"Remove TripleO job","message":"Remove TripleO job\n\nThe TripleO job has been in the experimental queue but has never get\nany actual attention. Also, the job is not used in Ceilometer.\n\nConflicts:\n\t.zuul.yaml\n\nBackport note:\nstable/xena and stable/yoga were skipped because the tripleo job has\nbeen removed from these branches by different commits.\n\nChange-Id: I074ef0ad4578abde73ae82efb18bc9edbceb7e81\n(cherry picked from commit 2eb0ae38fc54e9ab24dcfcb8528eb7dd340debfc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b4cae54cd9bbd1ce3cf2bec2ad8d28ffb9ea381e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b4cae54cd9bbd1ce3cf2bec2ad8d28ffb9ea381e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
