)]}'
{"id":"openstack%2Faodh~867592","triplet_id":"openstack%2Faodh~stable%2Fvictoria~I074ef0ad4578abde73ae82efb18bc9edbceb7e81","project":"openstack/aodh","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-19 07:58:23.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I074ef0ad4578abde73ae82efb18bc9edbceb7e81","subject":"Remove TripleO job","status":"MERGED","created":"2022-12-14 03:31:11.000000000","updated":"2022-12-19 09:54:33.000000000","submitted":"2022-12-19 09:52:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867592","meta_rev_id":"f4927df67395642e4d682fd9b91644472f58e901","_number":867592,"virtual_id_number":867592,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-19 09:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-12-14 07:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-12-14 07:02:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-14 04:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-14 07:02:58.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"cdd1ccb2709f39f749857707d6ce6f332e064eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-14 03:31:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be3420b453c0ab8adc9ba5d61bf28c2896a90232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 04:32:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7a8f1244c3849718d26eea75eb30886\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebb74efd58a4871a310bfdea1e9d720 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5c13826166e4346837e07cc5baece39 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5076c8538564106bed42ddce06171b6 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caac0c63adcd4724b2940077e42866db : FAILURE in 5m 18s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a8b41e2f893e4a0aa27aa468445654d9 : SUCCESS in 51m 54s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3189f372725247feb3f7fb01c2f4e912 : SUCCESS in 46m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b7429c96c68b4016ae511c7005e41251 : SUCCESS in 51m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/75c8b700285f48d7abda2062bc17a5af : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f65495d7523ff8a7c300d1650c4ef1dc85062658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-14 05:36:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d8847ca713e200ca580c39f61fe21a4861ee8f8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 06:26:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ee380effee43bd9f5de9b6e267372a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e36f664c79104b98b57fecd5b4df30dc : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7360203fed9743d2b53a4465759170a0 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8305286886784781ba49c046563dd559 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6cd5823dc54079bbaddaa5dcab308e : SUCCESS in 7m 36s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/732fa3932978435ab9691a2635b9db8f : SUCCESS in 41m 52s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0030a5197fc940168f59025bf81d0e49 : SUCCESS in 40m 29s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3795450219894f169a00022711a9cee4 : SUCCESS in 47m 18s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/31cc5f9518dc4c67a1f440874e7b13fb : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":2},{"id":"82f45851dec444cc9133fb10630b1e5d80bb0647","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-14 07:02:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e62d0d7f4602775d75a26a0fd4b0eb5ce67e8c65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 06:46:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d18103476afe63fede081dbabc42a9e85b3af3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 07:44:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbba5aaac84648b0ba126a55229520d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a5546d8948a4a77bd76fd1dccab6792 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a5c3f23b6f4e779c79b95c690bd28b : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/456082bc5bc24eb5857b3c8f467def7d : FAILURE in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a08ff6a860b4aadb712d21a6de65cd2 : SUCCESS in 6m 39s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a6e85fcbd0cb47c8929a24e603c446b9 : SUCCESS in 52m 41s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4f17c0e75b624365aa1567d9f287935f : SUCCESS in 28m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/fe9f574478454514b4d55e82f9a601ad : SUCCESS in 32m 39s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a807d1a7dd324dad9a6dbcc083d8b404 : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":2},{"id":"cac9f1d5830606200cd8b65914e4aaf1ef08e843","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-19 07:58:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6590bcc91e82f9a78cd5af1f462eecdfca80b386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 08:51:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01cd751a02314a67ba13203d6f52a76a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79f48dac0ff4d4daf224f18771401eb : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a06e199b593643628b6df2ad10a6cf9b : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f32e620b1864f46a679ddb56474e274 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f7e31c27e34520844a3406af77d4d6 : SUCCESS in 7m 57s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/67aff7d5f5ad4abbaa1224cd0648ad39 : SUCCESS in 41m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/333be5614f7c45c28ca04aeb94523ff4 : SUCCESS in 37m 28s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/89b13aecb68147d3a531d6e8a6369edd : SUCCESS in 51m 14s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f4ab382481a3473a804e86f5028646d8 : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":2},{"id":"26eab4e4ec8300b806754974bbebd46e95492173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 08:52:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb4a45401b87dac046f676f57d4a1a3785bc4cb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 09:52:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/968fc5ef2f3f454294e970397c21faff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62085153751146a095fde02660908198 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3286155e508e4d479d2fdcf56a60e605 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19c80fd213d44fb4b39b6dde54fe5f4c : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8c8f687c8648e692ac8cc49f0ab367 : SUCCESS in 7m 46s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e00f0d03696f4e919bef5c6c60966a42 : SUCCESS in 34m 09s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2ec84c4b7de04cd7a05c0eb92afd1e26 : SUCCESS in 42m 29s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/692d1f9f76894c31aa04a72670311e54 : SUCCESS in 54m 43s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4a42de93213b4c2586084af03bbe1dcb : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":2},{"id":"613d53ec510af63b5f1b22f4c0dea33eeac600b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 09:52:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f4927df67395642e4d682fd9b91644472f58e901","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 09:54:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba7484ec1494cf2aa21737c5e69d7cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada56b99124b4146bbfff43a530a2f31 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b","revisions":{"a892c2408da1281155a64f151a11f6dcadf81e67":{"kind":"REWORK","_number":1,"created":"2022-12-14 03:31:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/867592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/92/867592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/867592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/867592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/867592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/92/867592/1"}}},"commit":{"parents":[{"commit":"028561741e125f1b343ae5bf1021df1012e92005","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/028561741e125f1b343ae5bf1021df1012e92005"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-26 02:03:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-14 03:31:00.000000000","tz":540},"subject":"Remove TripleO job","message":"Remove TripleO job\n\nThe TripleO job has been in the experimental queue but has never get\nany actual attention. Also, the job is not used in Ceilometer.\n\nConflicts:\n\t.zuul.yaml\n\nBackport note:\nstable/xena and stable/yoga were skipped because the tripleo job has\nbeen removed from these branches by different commits.\n\nChange-Id: I074ef0ad4578abde73ae82efb18bc9edbceb7e81\n(cherry picked from commit 2eb0ae38fc54e9ab24dcfcb8528eb7dd340debfc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a892c2408da1281155a64f151a11f6dcadf81e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a892c2408da1281155a64f151a11f6dcadf81e67"}]},"branch":"refs/heads/stable/victoria"},"85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-14 05:36:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/867592/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/92/867592/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/867592/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/867592/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/867592/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/92/867592/2"}}},"commit":{"parents":[{"commit":"7e73e6111e7173604bcb1ef28a285e4d47fbc72f","subject":"Replace oslo_utils.fnmatch with fnmatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7e73e6111e7173604bcb1ef28a285e4d47fbc72f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-26 02:03:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-14 05:36:13.000000000","tz":0},"subject":"Remove TripleO job","message":"Remove TripleO job\n\nThe TripleO job has been in the experimental queue but has never get\nany actual attention. Also, the job is not used in Ceilometer.\n\nConflicts:\n\t.zuul.yaml\n\nBackport note:\nstable/xena and stable/yoga were skipped because the tripleo job has\nbeen removed from these branches by different commits.\n\nChange-Id: I074ef0ad4578abde73ae82efb18bc9edbceb7e81\n(cherry picked from commit 2eb0ae38fc54e9ab24dcfcb8528eb7dd340debfc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
