)]}'
{"id":"openstack%2Faodh~867594","triplet_id":"openstack%2Faodh~stable%2Ftrain~I074ef0ad4578abde73ae82efb18bc9edbceb7e81","project":"openstack/aodh","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-22 01:14:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I074ef0ad4578abde73ae82efb18bc9edbceb7e81","subject":"Remove TripleO job","status":"MERGED","created":"2022-12-14 03:32:25.000000000","updated":"2022-12-22 03:26:09.000000000","submitted":"2022-12-22 03:24:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867594","meta_rev_id":"ab4079a9a50892f4ba4b1a8d482c4c4df65eac58","_number":867594,"virtual_id_number":867594,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-22 03:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-12-20 11:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-12-20 11:30:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-14 04:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 11:30:49.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"fa040c75f535ba8c06ef290a2dcd7e50614f58f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-14 03:32:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbbbce13705e2ecec0fc2fe8fbc704d4fda1fd5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 04:18:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1da08a18bf22465e83a76853b2e627eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d7852c89e0459f8559e20b9463b066 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/587f3cba40044c98925b077b6098f049 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/460bc51e83be4a7f9834564346a518b7 : SUCCESS in 7m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eee1629f0fdf4ff598288419a885ee4f : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d43ef0b4b34c62af40d83d977ded65 : SUCCESS in 7m 10s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/277aadc6f7aa4cd084fade8d4486b6fd : SUCCESS in 41m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/07a5bbb52357430cbfd8c6b4d2d9dfcb : SUCCESS in 28m 24s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4483df010fee44c99dcfa630f71854e7 : SUCCESS in 32m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9d5b4fe483024f81a3b0ab1c5702bc11 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d168edb0c3e98641ae5f6d3dd68e87e69a4f6de3","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-20 11:30:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67de505a5e4119adfaac09cb2901243eb9d7dbfd","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-20 11:30:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c9eb6f4c77373199d7b7ad17b3fa68e72106b18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 11:31:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"565e17ab9c415b62644901a80f3d008826c4e11c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 12:34:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0b5ebd950ed48689ad05df0a48e275d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/470bf294c3fa496a869adbc3d9a93000 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2082d1bcc194e419f8ac5d0f6b26915 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14edb5f38548435db5987a063565be09 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/867bdeaeadf14151abd9b0f36f95bad9 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2498edda8343b7bed56b670527ce4f : SUCCESS in 7m 14s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/62b18e8af9014853bd526817d1b300c3 : SUCCESS in 51m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e680b97fb6a64d8d84f7315992151ade : SUCCESS in 51m 24s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3d77df310c924c2c9652894c73417b0c : FAILURE in 58m 28s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2bb99f3b620b46988757c64cc2c159b5 : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5503ef7e2379b56f47eae375c894caa3b4f0a6c8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-22 01:14:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71331a8a63992470f437d3c913f96d0b91e7cb76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 02:16:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2178fc23de474e479b581f4d9e0e58d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06c8d6aea60b4c959f6276dbeeb67054 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ceb4420b6f7343fb872c46a4f1383de5 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e00f7c0ec0345198c559a93cc6f00a0 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a95f743f5fc24305b89fdbe1eb1e0278 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fb43101f1849cf85311a8b6b2d7ab0 : SUCCESS in 7m 06s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c5830e6e48294e4087a6e58f281e3b5e : SUCCESS in 51m 37s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7c59d597b465431f8e57960c68f67ea5 : SUCCESS in 48m 08s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/796484725cd540d9a2f313cf7e800dee : SUCCESS in 56m 05s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/090482f4101a41a4ad9a3aa09d846eef : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0bb32697e0b6279e0ac3beb9704886a31de36ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 02:17:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"001377bdae4bda313fe8563a42e12ad80eb8f925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 03:24:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f58b3ce5447b4d3ba44ffe4fa89cb35c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb6786b86b445f5b638f2e2003a3ace : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3abc88f9dab475497bfd82b60e533bb : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94f246f86b034317a10b800b7c39390e : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/624c0757fa524d83b260bac46ee569d2 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dbd55498d524f1e9a25ee4d9fba6ef8 : SUCCESS in 6m 42s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8c5902845c6a478b8e44d478e1159f29 : SUCCESS in 48m 24s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b1028e268c4d47c58aef619a30d24685 : SUCCESS in 45m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2e345ca1d211487c9b833fe42ad5dfdb : SUCCESS in 52m 51s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3435089ca0b54e929542e10ee4db3314 : SUCCESS in 44m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d365c794c35974e89164d997587c2841d6709030","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 03:24:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab4079a9a50892f4ba4b1a8d482c4c4df65eac58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 03:26:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627aba05d2594253b024750cb99b4de2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2082941d24dc4091bed44d038ce27774 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2024b22104f7a9a3ba5758667a556d217c13883","revisions":{"c2024b22104f7a9a3ba5758667a556d217c13883":{"kind":"REWORK","_number":1,"created":"2022-12-14 03:32:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/867594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/94/867594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/867594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/867594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/867594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/94/867594/1"}}},"commit":{"parents":[{"commit":"5415b96d1e58615fa80fee2eb74dc9cda90d8e2a","subject":"Use config options when creating keystone session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5415b96d1e58615fa80fee2eb74dc9cda90d8e2a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-26 02:03:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-14 03:32:16.000000000","tz":540},"subject":"Remove TripleO job","message":"Remove TripleO job\n\nThe TripleO job has been in the experimental queue but has never get\nany actual attention. Also, the job is not used in Ceilometer.\n\nConflicts:\n\t.zuul.yaml\n\nBackport note:\nstable/xena and stable/yoga were skipped because the tripleo job has\nbeen removed from these branches by different commits.\n\nChange-Id: I074ef0ad4578abde73ae82efb18bc9edbceb7e81\n(cherry picked from commit 2eb0ae38fc54e9ab24dcfcb8528eb7dd340debfc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c2024b22104f7a9a3ba5758667a556d217c13883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c2024b22104f7a9a3ba5758667a556d217c13883"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
