)]}'
{"id":"openstack%2Faodh~868170","triplet_id":"openstack%2Faodh~stable%2Fvictoria~I4f631d224404460138f4050b1b981d577b592544","project":"openstack/aodh","branch":"stable/victoria","topic":"bug/1946793-stable/xena-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2022-12-20 10:04:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f631d224404460138f4050b1b981d577b592544","subject":"gnocchi: Use Dynamic Aggregates API","status":"MERGED","created":"2022-12-20 05:12:10.000000000","updated":"2022-12-20 10:05:41.000000000","submitted":"2022-12-20 10:04:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868170-bug/1946793-stable/xena-stable/wallaby-stable/victoria","cherry_pick_of_change":863798,"cherry_pick_of_patch_set":2,"meta_rev_id":"53d4b904feb5b39a01cf799ca0ff0ca3990c39d4","_number":868170,"virtual_id_number":868170,"owner":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 10:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-12-20 09:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-12-20 09:14:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 05:12:10.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"CC"},{"updated":"2022-12-20 06:31:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 09:14:57.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"f6f07e3f6f26e5cfdcc5429ec6b2486d1d950610","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-20 05:12:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* aodh/tests/unit/evaluator/test_composite.py\n* aodh/tests/unit/evaluator/test_gnocchi.py","accounts_in_message":[],"_revision_number":1},{"id":"2058ea6c249f4945489fb629935148d03881b9bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-20 05:17:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a51934c1cdd7c1b32e226fe23c267b41501bc68","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-20 05:17:35.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"5e18268a155530fc82f621ed6337f9adc01542d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-20 05:26:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"093866bc336b705ba7ab43361015d3f4fb5435e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-20 05:37:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a8c0e5d70db496e70e06ac63ecfd0740473a6d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 06:31:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5777bd43498c4acdaa2346b13c9bdd5c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba945e2b69f41e1a718c1db03a57124 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c44ab44d6247a0811bd7c313c8bd1e : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce79e7bf64864e889d360b1b02781aa7 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d3640b253d4c61885d19ffe05423de : SUCCESS in 5m 47s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8b9d54e7843a4692b67ce4be9e4b002c : SUCCESS in 30m 59s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/289bcf6647794666ae70b7a8da6b607f : SUCCESS in 47m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/38c0a13a5e73465d8119f71bdbb8e81a : SUCCESS in 30m 48s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8391f1be0d434c13b9d1c1719375ecca : SUCCESS in 48m 50s","accounts_in_message":[],"_revision_number":4},{"id":"60fbe08044d030334d092ac625a1170beaa592ad","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-20 09:14:57.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"86f60df0ee5e2ab6678c0f9236aa7f8ddb91139a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 09:15:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dcf6c27fd7f322b6626cbdab873482eb365455ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:04:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c1bd95e459b4c0eac72e654b86aa26f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd75dec9a6894fd0b0fe59ddba17af68 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/128094fa4999414fbec5802488838648 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b417848d9de248019b3af9bd0b9cb599 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3190961d6dee4c1da016ee08eebbb184 : SUCCESS in 9m 00s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/934211eac3f840f5ad33843d850bccfd : SUCCESS in 45m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/06b31150e9b74e51841bc214ce6977ac : SUCCESS in 42m 21s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/da8f221e4557459db883f84b503b3078 : SUCCESS in 40m 34s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/aa797319314e4b528f73af406aa127a0 : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":4},{"id":"9789737145b6605a61b8843d4ff907b7c466f979","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:04:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"53d4b904feb5b39a01cf799ca0ff0ca3990c39d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:05:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeaa7a531e7243d78aa40460be1f570c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86fc17a03624b11a4cde68154a9d6de : SUCCESS in 40s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4820d595bfe4c4280b77bcac7fd97660397bfe39","revisions":{"f12e7e3abbe1380e6f7e28aba50583e2ed57cee7":{"kind":"REWORK","_number":1,"created":"2022-12-20 05:12:10.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/70/868170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/70/868170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/70/868170/1"}}},"commit":{"parents":[{"commit":"85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-12-20 05:12:10.000000000","tz":0},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n(cherry picked from commit e6d55b1ea9ee852cc56002c865c9443b526f2f58)\n(cherry picked from commit 27ee81a37acf097f572ffce1c4cd72f9987e0520)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f12e7e3abbe1380e6f7e28aba50583e2ed57cee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f12e7e3abbe1380e6f7e28aba50583e2ed57cee7"}]},"branch":"refs/heads/stable/victoria"},"f25b0be61c355853ebede3097d9f35302c1507c2":{"kind":"REWORK","_number":2,"created":"2022-12-20 05:17:18.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/70/868170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/70/868170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/70/868170/2"}}},"commit":{"parents":[{"commit":"85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-12-20 05:17:14.000000000","tz":330},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n(cherry picked from commit e6d55b1ea9ee852cc56002c865c9443b526f2f58)\n(cherry picked from commit 27ee81a37acf097f572ffce1c4cd72f9987e0520)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f25b0be61c355853ebede3097d9f35302c1507c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f25b0be61c355853ebede3097d9f35302c1507c2"}]},"branch":"refs/heads/stable/victoria"},"7e2de953e14baf2bf10178577bb1e138022df4cf":{"kind":"REWORK","_number":3,"created":"2022-12-20 05:26:11.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/70/868170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/70/868170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/70/868170/3"}}},"commit":{"parents":[{"commit":"85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-12-20 05:25:58.000000000","tz":330},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n(cherry picked from commit e6d55b1ea9ee852cc56002c865c9443b526f2f58)\n(cherry picked from commit 27ee81a37acf097f572ffce1c4cd72f9987e0520)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7e2de953e14baf2bf10178577bb1e138022df4cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7e2de953e14baf2bf10178577bb1e138022df4cf"}]},"branch":"refs/heads/stable/victoria"},"4820d595bfe4c4280b77bcac7fd97660397bfe39":{"kind":"REWORK","_number":4,"created":"2022-12-20 05:37:33.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/70/868170/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/70/868170/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/868170/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/70/868170/4"}}},"commit":{"parents":[{"commit":"85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b","subject":"Remove TripleO job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/85571a65b8901bd7a5a063db6eba5a5f5e8b8f9b"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-02-18 09:51:41.000000000","tz":0},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-12-20 05:37:22.000000000","tz":330},"subject":"gnocchi: Use Dynamic Aggregates API","message":"gnocchi: Use Dynamic Aggregates API\n\nSwitch to using the Dynamic Aggregates API as the Metric Aggregation\nAPI is deprecated.\n\nWhen using the Dynamic Aggregates API, any aggregation using rates\ncan use the underlying base measures for the aggregation rather than\nthe rate, for example:\n\n    (aggregation rate:mean (metric cpu mean))\n\nThe tuple of data for each record returned via this API is encapsulated\nwith information about the aggregation used so adapt the sanitization\nfunction to deal with this and the formatting of the metrics measures\nAPI as well.\n\nChange-Id: I4f631d224404460138f4050b1b981d577b592544\nCloses-Bug: 1946793\n(cherry picked from commit 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8)\n(cherry picked from commit e6d55b1ea9ee852cc56002c865c9443b526f2f58)\n(cherry picked from commit 27ee81a37acf097f572ffce1c4cd72f9987e0520)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4820d595bfe4c4280b77bcac7fd97660397bfe39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4820d595bfe4c4280b77bcac7fd97660397bfe39"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
