)]}'
{"id":"openstack%2Faodh~868174","triplet_id":"openstack%2Faodh~stable%2Fussuri~Ief75ccdc0fe5c6a197fc05bc481a5be096447558","project":"openstack/aodh","branch":"stable/ussuri","topic":"bug/1974682-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2022-12-21 15:16:31.000000000","reason":"\u003cGERRIT_ACCOUNT_32240\u003e replied on the change","reason_account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}}},"hashtags":[],"change_id":"Ief75ccdc0fe5c6a197fc05bc481a5be096447558","subject":"Ignore Gnocchi API error when the metric is not yet created","status":"MERGED","created":"2022-12-20 10:06:39.000000000","updated":"2022-12-21 17:29:45.000000000","submitted":"2022-12-21 17:28:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868174-bug/1974682-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":868171,"cherry_pick_of_patch_set":3,"meta_rev_id":"60be5c5e8e14fe27c4ec710c5087e38cff572771","_number":868174,"virtual_id_number":868174,"owner":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-21 17:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-12-20 11:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-12-20 11:26:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 10:06:39.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-12-20 10:52:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 11:26:59.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"35f8a6ebd20b105160a6e6c4a739bfba9748eee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-20 10:06:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"19e17817e00a1b9b9b53fb8979fbb81b5454a597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:52:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00b8ea4ab746487b81675c07abfd803a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e98b2dde71745639cc7670187a70368 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f089bcc549a4c26b04a6e32acb67c23 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8d2177a624e4413bd8809c9af99e557 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2c1062191da41389c32ebdd27163f24 : SUCCESS in 7m 29s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4fd09542b9eb4514a5b214c4b86a8121 : SUCCESS in 30m 38s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3986ef790b2c4721bb400b32c7e718cf : SUCCESS in 34m 01s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/43bd5a068701469f90dc8463b77b79a1 : SUCCESS in 35m 46s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/98af24ef3b7f4f7ca39cb1e527863c6e : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7bb2be9ea227ac64ed1c72a1a65300c93e92f27b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-20 11:17:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6635840a9a5a91c8e6286363a92dcaea2311d030","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-12-20 11:26:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f76e728b62c00ae18b79b609fc066c36e94c3abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 11:50:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3863ef28d5d64ec3ba8a9e287d53fdb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18808e439dd41f2939c2d1f4625a2d4 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df16b62a898e4761aa4c35b7319dafac : SUCCESS in 7m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/057cf9efe97547a28dbae2400f117656 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d911b315ee4d88baa7d7cd1bd7c7d0 : SUCCESS in 7m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/eaf9dbc99a5c4c4abc850a9eb9b2317f : SUCCESS in 31m 15s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7ba0d26468f04863b25b19aadff793b1 : SUCCESS in 30m 10s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7ce132d876714045aea0af8fd38bea6a : SUCCESS in 32m 05s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/43541e7f07e0458e869e058b9bfb9d19 : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8f3bcb052841e6121cb897f2b36d946337c93f85","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-12-21 15:16:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"276c928a9d3cb4b144c0577a9bf99383917709ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 16:29:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cfc670f8e6b4b7f9d64fd32ca10e3fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed367c54de747bbb7078b8f372e5f3f : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e8ff42c4160400cb4d97d676d7fe119 : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ac2f1e58d9b44cf84f8eaa91e3b96e5 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d075bf8ab0bb4f6bac080c47dff3ffb0 : SUCCESS in 7m 41s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/08ee9887e6ba45ec8500beb8d4556e48 : SUCCESS in 53m 10s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/53ed38f48a2948509af1c004a9ef9d25 : SUCCESS in 1h 06m 14s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cea0322aef764586bffa6019518dc801 : SUCCESS in 51m 16s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2601a54867db4cf8891377036f1a3814 : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b443230f693e6a983b8dc598697b8092076f6258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 16:30:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93dc56f55c1b7d8a8844d4003e42f57a3ea1e4e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 17:28:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b526030c15fa44f5b09add36cb226425\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d62f6d4d9844f09179682e7fff5f77 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2728d018d72489c94ffcc821531397c : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a728f32ea481453daa72a8e32b8b2391 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1203dc146984f6f9b8eaddfea8d6991 : SUCCESS in 6m 23s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/fb140595406d4118adb5aeed472dfa99 : SUCCESS in 48m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3c75179a9a764c15a14aaa57a493fae9 : SUCCESS in 41m 14s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1b6290acbe8e4f568221f4afc87a4500 : SUCCESS in 52m 21s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/acc493c040da4b09bd9b14ce87ce4a34 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":2},{"id":"49689823165a5bbb7d0a76941176c8e7e4c250ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 17:28:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"60be5c5e8e14fe27c4ec710c5087e38cff572771","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 17:29:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17da3f8940854ae887247c83dd444be6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47fbbeb733ee43c18a45d063ee9f60b3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da0fd0ca4b90601ecf27c38b2b7b7aae3a516e90","revisions":{"7a9529542be719ebf0384217208b302b7cd05e09":{"kind":"REWORK","_number":1,"created":"2022-12-20 10:06:39.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/74/868174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/74/868174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/868174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/868174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/868174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/74/868174/1"}}},"commit":{"parents":[{"commit":"d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-30 04:33:06.000000000","tz":540},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-12-20 10:06:39.000000000","tz":0},"subject":"Ignore Gnocchi API error when the metric is not yet created","message":"Ignore Gnocchi API error when the metric is not yet created\n\nChange 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8 replaced usage of\nmetric aggregate API by Dynamic aggregate API.\n\nHowever the Dynamic aggregate API returns a different response code\n(not 404, but 400) in Gnocchi \u003c\u003d 4.4 and this is causing the direct API\nerror when the metric associated with the aggregation does not exist.\n\nThis change adds 400 to the ignored response codes so that Aodh api\ndoes not return error even when the associated metric is not yet\ncreated in Gnocchi.\n\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/868170\n\nCloses-Bug: #1974682\nChange-Id: Ief75ccdc0fe5c6a197fc05bc481a5be096447558\n(cherry picked from commit 214be56b3865d4d5bc30eaf0cb4b56bde9577f8c)\n(cherry picked from commit 2ad64b04e858565fb830ba207841b70edf5ff347)\n(cherry picked from commit 231f404ad13bac69964f19b9b1bfa5a42c1f977a)\n(cherry picked from commit 8454cac06a654bb1e4eea20e6a36417796a94d82)\n(cherry picked from commit 4eaa8d64ece760837243b89caf8e43baf033d975)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7a9529542be719ebf0384217208b302b7cd05e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7a9529542be719ebf0384217208b302b7cd05e09"}]},"branch":"refs/heads/stable/ussuri"},"da0fd0ca4b90601ecf27c38b2b7b7aae3a516e90":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-20 11:17:23.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/74/868174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/74/868174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/868174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/868174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/868174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/74/868174/2"}}},"commit":{"parents":[{"commit":"d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d7b4f91c2f2e9d6a18a10ee4b668f8f78890cdb3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-30 04:33:06.000000000","tz":540},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-12-20 11:17:23.000000000","tz":0},"subject":"Ignore Gnocchi API error when the metric is not yet created","message":"Ignore Gnocchi API error when the metric is not yet created\n\nChange 74eadfbd58359b7ebe9e1e40ae6b6ff245146bb8 replaced usage of\nmetric aggregate API by Dynamic aggregate API.\n\nHowever the Dynamic aggregate API returns a different response code\n(not 404, but 400) in Gnocchi \u003c\u003d 4.4 and this is causing the direct API\nerror when the metric associated with the aggregation does not exist.\n\nThis change adds 400 to the ignored response codes so that Aodh api\ndoes not return error even when the associated metric is not yet\ncreated in Gnocchi.\n\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/868173\n\nCloses-Bug: #1974682\nChange-Id: Ief75ccdc0fe5c6a197fc05bc481a5be096447558\n(cherry picked from commit 214be56b3865d4d5bc30eaf0cb4b56bde9577f8c)\n(cherry picked from commit 2ad64b04e858565fb830ba207841b70edf5ff347)\n(cherry picked from commit 231f404ad13bac69964f19b9b1bfa5a42c1f977a)\n(cherry picked from commit 8454cac06a654bb1e4eea20e6a36417796a94d82)\n(cherry picked from commit 4eaa8d64ece760837243b89caf8e43baf033d975)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/da0fd0ca4b90601ecf27c38b2b7b7aae3a516e90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/da0fd0ca4b90601ecf27c38b2b7b7aae3a516e90"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
