)]}'
{"id":"openstack%2Faodh~872419","triplet_id":"openstack%2Faodh~master~I5941fe03277368f789a0e07d823ececdf86ed37e","project":"openstack/aodh","branch":"master","topic":"story/2010559","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2023-03-10 10:31:10.000000000","reason":"\u003cGERRIT_ACCOUNT_32240\u003e replied on the change","reason_account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2023-03-10 12:04:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5941fe03277368f789a0e07d823ececdf86ed37e","subject":"Add oslo.policy.enforcer entry point","status":"MERGED","created":"2023-02-01 15:16:36.000000000","updated":"2023-03-10 12:05:07.000000000","submitted":"2023-03-10 12:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872419-story/2010559","meta_rev_id":"657a4c42df1cfb63955ded74128d1883d8b9c3f9","_number":872419,"virtual_id_number":872419,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-10 12:04:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"all":[{"value":2,"date":"2023-03-10 10:31:10.000000000","_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":2,"date":"2023-03-03 14:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":1,"date":"2023-03-10 10:56:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-01 16:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-03 14:40:43.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-03-07 09:08:32.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"}],"messages":[{"id":"1deb557a7279c5f62255a67ed32566441a4ec4aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-02-01 15:16:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"535a29eec92e53d6a48f44e96c6b2d3458f131da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:40:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a79f9ec6a356472abf23bad13433b172\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9c2b8647957941119967bc1fc85cc062 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/286249f507f949e79c357cf5eb22624b : FAILURE in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d80acb4ab7f46058a3d970365d2bf26 : FAILURE in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adbbb03a6cbb4e5d90d0f3af306c1a25 : FAILURE in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74610ac4ba1c484c9d68ba4af4e750cf : SUCCESS in 6m 47s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/59c1637317004a238c7e857a384c5e1e : FAILURE in 46m 55s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ec687d1eff814ee1b3f051e645dfec40 : FAILURE in 43m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9936e10326c74c0d8541d497744010ed : FAILURE in 49m 50s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f1c140a54dd146ba8f2db2dc1c41f373 : FAILURE in 49m 26s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0d98bcac09474e15b89993e7adb7add2 : FAILURE in 54m 16s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/74d3d28fe6e245ddbe9fc7622281a660 : FAILURE in 58m 06s","accounts_in_message":[],"_revision_number":1},{"id":"12ecdbea85d66d71085102a5cfeeee2dabd7fbff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-02-02 15:55:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e58d92201b7d5d81fe4bfc3bdda0491d943fd4a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 17:03:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4620a20d73a488fb8b0655f2f1370d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9243454f10e5472f9aa7a54412173a58 : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad340db2863f4f76b0a4eede7697e4e0 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84581ff4480a4a11949efef4103ee876 : SUCCESS in 8m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb9484105bbb4362856b6f9b9108164c : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e00635535e46529628013a22b630b5 : SUCCESS in 7m 41s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9f898a277401429c878595a09aec9c00 : SUCCESS in 42m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/395e424730d946508951ea43d0d9ddc7 : SUCCESS in 36m 03s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6986e2febd6043769ca7cae2eb89b94f : SUCCESS in 32m 13s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3254a63df2d84f188be6eb0f9f46be58 : SUCCESS in 1h 01m 37s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/2a5ffc9d44ad4a569f25bf38b2d101d6 : SUCCESS in 51m 05s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/8d72dc665b5f4dc6bb8bb79247736847 : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0791e4be7be26b5aa53c381e1d69b2677accfa5b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-03-03 14:40:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"50fc93e7723c41ce83ea2f14ac1ae5b9fb6f3388","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-03-10 10:31:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed79ad385497db48489488994246e7322f916be7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-03-10 10:56:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e017760ea426b57ac48d5316ec019199be613e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 10:57:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f55e908a3dec257046a2d33a9a8138474891614","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:04:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876073acf7494dec8b17324e03a35a8a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d49e31b3512548b09d96673573caf19b : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fc78e4c3134b2caedc517cb82eed00 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f8b36bb46fd46bba2dee639673e1715 : SUCCESS in 8m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bff93e9173e414f80d9606e49bdcf13 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7171c0d5ed84a388e092e5cf4425a8e : SUCCESS in 9m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/fc1bec55dc9a4d679ff7857aec212feb : SUCCESS in 37m 02s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/550baaabdeb049898146b70d2865864e : SUCCESS in 36m 17s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/346f0d14783e42c28cf68210c765cc72 : SUCCESS in 41m 56s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8591b349d7bd497da967f502d6b6b38e : SUCCESS in 1h 00m 52s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/2ed2e42fe9eb441db395fc85d0dd1de9 : SUCCESS in 48m 08s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/565ca1851630401ba9f34a836e9835ef : SUCCESS in 45m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c2f7c1411fed20fc1c876baf8792cf40db1d06e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:04:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"657a4c42df1cfb63955ded74128d1883d8b9c3f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:05:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26daf42b8d82420da7b347b79a091b9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83df26821378463688ccf1315751e99d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9d9840023e1fac2ed51754a9e96f0619c0fa960f","revisions":{"f9277dffe447bf1eba3bd7ac0ba900b5f0b2052f":{"kind":"REWORK","_number":1,"created":"2023-02-01 15:16:36.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/19/872419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/19/872419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/872419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/872419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/872419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/19/872419/1"}}},"commit":{"parents":[{"commit":"aa91a3b6141dd3177d581c5517ae9fe37e75ab4c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aa91a3b6141dd3177d581c5517ae9fe37e75ab4c"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-02-01 13:42:41.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-02-01 13:57:24.000000000","tz":120},"subject":"Add oslo.policy.enforcer entry point","message":"Add oslo.policy.enforcer entry point\n\nthis will allow various oslo.policy scripts like\noslopolicy-policy-generator to work with Aodh.\n\nChange-Id: I5941fe03277368f789a0e07d823ececdf86ed37e\nStory: 2010559\nTask: 47247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f9277dffe447bf1eba3bd7ac0ba900b5f0b2052f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f9277dffe447bf1eba3bd7ac0ba900b5f0b2052f"}]},"branch":"refs/heads/master"},"9d9840023e1fac2ed51754a9e96f0619c0fa960f":{"kind":"REWORK","_number":2,"created":"2023-02-02 15:55:14.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/19/872419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/19/872419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/872419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/872419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/19/872419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/19/872419/2"}}},"commit":{"parents":[{"commit":"aa91a3b6141dd3177d581c5517ae9fe37e75ab4c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aa91a3b6141dd3177d581c5517ae9fe37e75ab4c"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-02-01 13:42:41.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","date":"2023-02-02 15:55:14.000000000","tz":0},"subject":"Add oslo.policy.enforcer entry point","message":"Add oslo.policy.enforcer entry point\n\nthis will allow various oslo.policy scripts like\noslopolicy-policy-generator to work with Aodh.\n\nChange-Id: I5941fe03277368f789a0e07d823ececdf86ed37e\nStory: 2010559\nTask: 47247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9d9840023e1fac2ed51754a9e96f0619c0fa960f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9d9840023e1fac2ed51754a9e96f0619c0fa960f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
