)]}'
{"id":"openstack%2Faodh~879143","triplet_id":"openstack%2Faodh~master~I225514a2deece0affaeaf9fc574862254d17cd45","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"I225514a2deece0affaeaf9fc574862254d17cd45","subject":"Remove unused debtcollector","status":"ABANDONED","created":"2023-03-31 09:18:13.000000000","updated":"2024-01-24 03:37:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"12cd5e884f27fb23ee105d26ff60dd1db186120b","_number":879143,"virtual_id_number":879143,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-31 09:32:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 09:32:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fc59e94a65dd5133cf42b6865063505be2a4e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-31 09:18:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1639ffa784f4dc8e7de10c3030e716327f7e431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-31 09:19:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"73e6e2a76356ac5f404292892deec615160127f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 09:32:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45f1a51d88e44e648ed4033bdb9b8efc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0576bbcc0aa495ea318560aa55206bf : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a02248f854849a1b503b2fdf1ffc7f5 : SUCCESS in 7m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a111add5347344ab8927ddcb088802b6 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42fbabe9152a4fd0a3c3706c62fbf53e : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":2},{"id":"12cd5e884f27fb23ee105d26ff60dd1db186120b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-24 03:37:22.000000000","message":"Abandoned\n\nthis was done by 6c032153bff7af86f0dfdf5a3008f90f6a842e4a","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e7fdbfb9c1b55ded563e5237c00e4fece7c6588","revisions":{"7c1a0cde87bc9cbf699b7d47622c185a81496fef":{"kind":"REWORK","_number":1,"created":"2023-03-31 09:18:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/879143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/43/879143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/43/879143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/43/879143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/43/879143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/43/879143/1"}}},"commit":{"parents":[{"commit":"bb36bb618e11ce463990c61a601ae85bbf5425c7","subject":"Merge \"Add oslo.policy.enforcer entry point\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bb36bb618e11ce463990c61a601ae85bbf5425c7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-31 09:17:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-31 09:17:50.000000000","tz":540},"subject":"Remove unused debtcollector","message":"Remove unused debtcollector\n\nThis library is no longer used by any code in this repository.\n\nChange-Id: I225514a2deece0affaeaf9fc574862254d17cd45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7c1a0cde87bc9cbf699b7d47622c185a81496fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7c1a0cde87bc9cbf699b7d47622c185a81496fef"}]},"branch":"refs/heads/master"},"5e7fdbfb9c1b55ded563e5237c00e4fece7c6588":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-31 09:19:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/879143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/43/879143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/43/879143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/43/879143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/43/879143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/43/879143/2"}}},"commit":{"parents":[{"commit":"bb36bb618e11ce463990c61a601ae85bbf5425c7","subject":"Merge \"Add oslo.policy.enforcer entry point\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bb36bb618e11ce463990c61a601ae85bbf5425c7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-31 09:17:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-31 09:18:31.000000000","tz":540},"subject":"Remove unused debtcollector","message":"Remove unused debtcollector\n\nThis library is no longer used by any code in this repository.\n\nThe library was added when \u0027threshold\u0027 alarm type was deprecated[1],\nbut we missed removing it when we removed the deprecated feature[2].\n\n[1] 69cfeca3627d74b12f38ca463af0b5ea519f315d\n[2] b53d862bb5f9fc6528d13de3006f8bb0f66d14f2\n\nChange-Id: I225514a2deece0affaeaf9fc574862254d17cd45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5e7fdbfb9c1b55ded563e5237c00e4fece7c6588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5e7fdbfb9c1b55ded563e5237c00e4fece7c6588"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
