)]}'
{"id":"openstack%2Faodh~881808","triplet_id":"openstack%2Faodh~master~Iaff14e4b004e8562b41f8c94e40c2ee595884f6e","project":"openstack/aodh","branch":"master","topic":"requirements","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-05-10 06:02:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaff14e4b004e8562b41f8c94e40c2ee595884f6e","subject":"Drop tenacity cap to align with global_requirements","status":"MERGED","created":"2023-04-28 10:26:08.000000000","updated":"2023-05-10 06:03:10.000000000","submitted":"2023-05-10 06:02:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881808-requirements","meta_rev_id":"ca37d88d94792bd02a256897c7469a4831d86ded","_number":881808,"virtual_id_number":881808,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 06:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"all":[{"value":2,"date":"2023-05-10 04:35:06.000000000","_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":2,"date":"2023-04-28 12:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"all":[{"value":1,"date":"2023-05-10 05:48:31.000000000","_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-28 10:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-28 12:52:04.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-05-10 04:35:06.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"}],"messages":[{"id":"5bb0a79ff90d54b6ed82477bb92a4d17b400e7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-04-28 10:26:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b82a54d8634647b39594c794a7731fbbbd7d0f1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 10:51:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e471136230764c91b93264cc59396a42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38baa18b057e4712854e92a3565e3fa4 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/376eb305bebe454496590e70dc878394 : SUCCESS in 7m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcd35d58fd8c4b9da51f8f8938a08ea4 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96f5ac537864f658c721288a4c4a3b1 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"86d12b3f82044f018e8790f03a349ae0a2f577fc","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-04-28 12:52:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ae704fa3fd363e5ce41277747b131b297217661","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-05-10 04:35:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd2fb71c287448120a0f36c77a76db19cc152d16","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-05-10 05:48:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc9121e15e1343089da0e7b30ec37c4ea5715ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 05:48:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c902ff6ff9d7f3f6d8e0de28754670a3b82a1b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 06:02:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8583708da2744168a852c623be7293e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd2f58dc9f740cfacfddce663bd07e3 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f5d0b79347418fb54c295d55c1f17a : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e309fb9739b04cf4ab7fa40a19b164c5 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26a57ea1ea654afd98486c4504ab41ef : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d42ec03eff714f4daedacb7bc88b1ba7 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5840f787b4da2186b2c1b109bc0da3d149150967","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 06:02:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca37d88d94792bd02a256897c7469a4831d86ded","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 06:03:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43d355b2002b4fc0b7aea60b8edccf54\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3f446d392d4052a9a7890dcfe888cf : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0782615601b96c1b790bd7aaf7d41f6b9a2611ee","revisions":{"0782615601b96c1b790bd7aaf7d41f6b9a2611ee":{"kind":"REWORK","_number":1,"created":"2023-04-28 10:26:08.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/08/881808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/08/881808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/08/881808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/08/881808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/08/881808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/08/881808/1"}}},"commit":{"parents":[{"commit":"bb36bb618e11ce463990c61a601ae85bbf5425c7","subject":"Merge \"Add oslo.policy.enforcer entry point\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bb36bb618e11ce463990c61a601ae85bbf5425c7"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-04-28 10:20:46.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-04-28 10:24:46.000000000","tz":60},"subject":"Drop tenacity cap to align with global_requirements","message":"Drop tenacity cap to align with global_requirements\n\nThe cap does not seem to be needed anymore.\n\nChange-Id: Iaff14e4b004e8562b41f8c94e40c2ee595884f6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0782615601b96c1b790bd7aaf7d41f6b9a2611ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0782615601b96c1b790bd7aaf7d41f6b9a2611ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
