)]}'
{"id":"openstack%2Faodh~882412","triplet_id":"openstack%2Faodh~master~Ia8bb7c1f17a8090c4932541ac3eed3a8a73f3252","project":"openstack/aodh","branch":"master","topic":"requirements","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2023-05-10 15:21:33.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-05-10 15:21:33.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2023-05-10 13:20:28.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia8bb7c1f17a8090c4932541ac3eed3a8a73f3252","subject":"Remove \"test\" extra deps from setup.cfg","status":"MERGED","created":"2023-05-05 10:04:45.000000000","updated":"2023-05-10 15:23:02.000000000","submitted":"2023-05-10 15:21:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882412-requirements","meta_rev_id":"03899fab1e2e7e252f960f23aa3ed41019ba38db","_number":882412,"virtual_id_number":882412,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 15:21:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2023-05-10 13:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2023-05-10 15:04:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-05 10:15:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-08 12:34:40.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-05-10 05:15:46.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"CC"}],"messages":[{"id":"cc491670ac45aa569d235b81bf5b10a4fcef1994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-05-05 10:04:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb629ae70b75e368332dfdc29dcd03b95e0093e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:15:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/018f186bec56493e960a8b76796eea6e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e3d9f367cb754a0eb8a0dd21cded2deb : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f23e6279feb74242aaef9142cd39dc19 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a67e6289c6945128e3b5f1b228cf0a6 : FAILURE in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f38c4251904e40d49212a8fb00a82440 : FAILURE in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ee9190dfcb493c8a37dfbba16edafa : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e949ad8b21bc2caaa1411018867d0b81d52c6fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-05-05 11:27:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c3c0f0c6b13bc969979d95181d2b8f58650e8ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 11:41:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fee608baf16425c9630608ac8c28222\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f62b810a73ac404897ae64c4805495ea : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b0be33a7124af99075839f4d4119e8 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e4d60c268804cccad388adef5b9a497 : SUCCESS in 7m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff34667231e346a38eab225f11c85f81 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36bcafb2271c4cdaa84ed4e1468fc55d : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ebb737432704bd3a7d03ef3010c38a0 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a886aa0d02f9c8916ba792f3d68081b4755067f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-05-05 12:18:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a8dbfcf046e1327fa56148a1538b089b01f45df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 12:35:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91690178407a40c18766223fcdc9716b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4025cdc415754d06b55e1eca86202d48 : SUCCESS in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13466387a4754cb1b0ff046967c64df1 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09aa4ea53d3f41168a103d0e1c63176b : SUCCESS in 7m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29cf8ceeeabb47689c622a811660de6f : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd33dd89fe8a4f6cb327b8fb00160819 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ceb657e9ee74f99a0ad70e89eeb8e85 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"344458e3e21b83f6855821e2d47d8bf4d9ed4c9b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-05-08 12:34:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d6b8067eeeeac025a9be8e70b99e0ce6748fe2d0","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-05-10 05:15:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"35e68b36f8e356ceb6b14f99627ca3d42a2db944","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-05-10 05:52:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aabfcb250b60f8983f0f02f39e3654352863f48e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-05-10 12:31:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ba113855bf9234577b01e2a05312eb9edd483fc4","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-05-10 12:32:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8a73803e668da114e9518514d0a9cc4ee4f4ab69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 12:45:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acaabea3f33e4c3db218e0808b3f9596\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6bd3fcacaf3a4cb8aab38f6748d06096 : SUCCESS in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d78a6c48325b4f87891c5d363b0061aa : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c38f2256c1a4b56ba3032c77fee4dbf : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17999205f5544e23a06859531f32dee2 : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63e6a0c42a384a579ffcf1e126ae9749 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da74b883026f45b5a8af205c70866b12 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49d7a45e0742431787ce967a60c3f8bc : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":4},{"id":"d7fd8aa915271e0f76e98a9cd33bc0cfc679469c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-05-10 13:20:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c145e8e5b747ed8756e88fd97b6725906abb5fbc","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-05-10 15:04:51.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5938eeade27ff97c10dfb54c4e229cab48505ed4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:05:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87473e4778545310217e2214350d442f5931bf73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:21:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da9a05a30b642029ee12e0af102a3d0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/44bad4c345424446949703e63a965a69 : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c25f3b58c3c4c58b9e0748b1a714c46 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a58f9e794bae4bffae729603f41abbd8 : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cd97d8abafc4821a6f1000a6c56b08f : SUCCESS in 8m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/170904fab2b64925b4b648816324f552 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9286732c0ab4fd5b5a34a3bfd4e2152 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72575b1ff008461aa749c004f07289d9 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":4},{"id":"767dc0ba2c1d7b666e8a2e8529ba11803a88a244","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:21:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"03899fab1e2e7e252f960f23aa3ed41019ba38db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:23:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c71fce014ca346758f2a60d39106bdb7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa89e011437455280d1933daa3efb3e : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3695a290a2314f18ab4539710c18e19b : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7ae3fd78c394999ff51aa891837d9151ddc765e9","revisions":{"26d19dcfef8575b598b1b9dc8783ebafda3e0821":{"kind":"REWORK","_number":1,"created":"2023-05-05 10:04:45.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/12/882412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/12/882412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/12/882412/1"}}},"commit":{"parents":[{"commit":"bb36bb618e11ce463990c61a601ae85bbf5425c7","subject":"Merge \"Add oslo.policy.enforcer entry point\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bb36bb618e11ce463990c61a601ae85bbf5425c7"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-05-05 10:04:38.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-05-05 10:04:38.000000000","tz":60},"subject":"DNM: testing the jobs","message":"DNM: testing the jobs\n\nChange-Id: Ia8bb7c1f17a8090c4932541ac3eed3a8a73f3252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/26d19dcfef8575b598b1b9dc8783ebafda3e0821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/26d19dcfef8575b598b1b9dc8783ebafda3e0821"}]},"branch":"refs/heads/master"},"18e13e777e2cdfbaffe16098eb4da716f0134962":{"kind":"REWORK","_number":2,"created":"2023-05-05 11:27:47.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/12/882412/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/12/882412/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/12/882412/2"}}},"commit":{"parents":[{"commit":"0782615601b96c1b790bd7aaf7d41f6b9a2611ee","subject":"Drop tenacity cap to align with global_requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0782615601b96c1b790bd7aaf7d41f6b9a2611ee"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-05-05 10:04:38.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-05-05 11:25:22.000000000","tz":60},"subject":"DNM: testing the jobs","message":"DNM: testing the jobs\n\nChange-Id: Ia8bb7c1f17a8090c4932541ac3eed3a8a73f3252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/18e13e777e2cdfbaffe16098eb4da716f0134962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/18e13e777e2cdfbaffe16098eb4da716f0134962"}]},"branch":"refs/heads/master"},"9896688226534c9fa0ddeb67e3e4c83e23f3392f":{"kind":"REWORK","_number":3,"created":"2023-05-05 12:18:01.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/12/882412/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/12/882412/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/12/882412/3"}}},"commit":{"parents":[{"commit":"0782615601b96c1b790bd7aaf7d41f6b9a2611ee","subject":"Drop tenacity cap to align with global_requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0782615601b96c1b790bd7aaf7d41f6b9a2611ee"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-05-05 10:04:38.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-05-05 12:17:24.000000000","tz":60},"subject":"Remove \"test\" extra deps from setup.cfg","message":"Remove \"test\" extra deps from setup.cfg\n\nUtilizing the \"test-requirements.txt\" for the test requirements\nand building tox \u0027testenv\u0027 with pifpaf and gnocchi rather than\nassuming they are actual lib dependencies of Aodh tests.\n\nChange-Id: Ia8bb7c1f17a8090c4932541ac3eed3a8a73f3252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9896688226534c9fa0ddeb67e3e4c83e23f3392f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9896688226534c9fa0ddeb67e3e4c83e23f3392f"}]},"branch":"refs/heads/master"},"7ae3fd78c394999ff51aa891837d9151ddc765e9":{"kind":"REWORK","_number":4,"created":"2023-05-10 12:31:37.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/12/882412/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/12/882412/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/12/882412/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/12/882412/4"}}},"commit":{"parents":[{"commit":"0782615601b96c1b790bd7aaf7d41f6b9a2611ee","subject":"Drop tenacity cap to align with global_requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0782615601b96c1b790bd7aaf7d41f6b9a2611ee"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-05-05 10:04:38.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-05-10 12:30:51.000000000","tz":60},"subject":"Remove \"test\" extra deps from setup.cfg","message":"Remove \"test\" extra deps from setup.cfg\n\nUtilizing the \"test-requirements.txt\" for the test requirements\nand building tox \u0027testenv\u0027 with pifpaf and gnocchi rather than\nassuming they are actual lib dependencies of Aodh tests.\n\nChange-Id: Ia8bb7c1f17a8090c4932541ac3eed3a8a73f3252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7ae3fd78c394999ff51aa891837d9151ddc765e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7ae3fd78c394999ff51aa891837d9151ddc765e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
