)]}'
{"id":"openstack%2Faodh~891562","triplet_id":"openstack%2Faodh~master~Ic094df537b5f7333be38c249cdec59a10d2e2660","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2024-01-09 05:24:44.000000000","reason":"\u003cGERRIT_ACCOUNT_32240\u003e replied on the change","reason_account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-12-15 03:55:51.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-09 06:46:34.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2023-12-15 07:26:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic094df537b5f7333be38c249cdec59a10d2e2660","subject":"Dump loaded options during start up","status":"MERGED","created":"2023-08-16 05:19:41.000000000","updated":"2024-01-09 06:47:29.000000000","submitted":"2024-01-09 06:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891562","meta_rev_id":"5c93c62d1b3140838922fa73e4bc5202d02f0fe3","_number":891562,"virtual_id_number":891562,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-09 06:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"all":[{"value":2,"date":"2024-01-09 05:24:44.000000000","_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":2,"date":"2023-12-15 07:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"all":[{"value":1,"date":"2024-01-09 05:24:44.000000000","_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-16 06:09:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-15 03:55:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"},{"updated":"2023-12-15 03:55:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2023-12-15 03:55:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"9deebf68c2686253c29469f4d06ae4570a932509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 05:19:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"484843cae0d27d6f3568b8162cb2a92790269d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 06:09:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8a2b2551d854e3fbec06e02a5cb70bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c9b4596c6c4572bde547309f8e918f : FAILURE in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e028022b395049dfbfe947b26eaf19ba : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15ffb98ab59c42e88f521ed519e01e93 : SUCCESS in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10b18cff30144a34a6ed4dccc9d68705 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/795188e04109496dbf72069a54cb1f8a : SUCCESS in 7m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6044a9cd899c424b95ccfb7648c8130c : FAILURE in 29m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8f3cfbd9e62a41a3bb5fe13f128ea2a4 : FAILURE in 24m 55s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b0ff10cff30a4d57a93e5ba35a25d6da : FAILURE in 26m 54s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3fbbbbf338c641d9b4dcab5b44f7422d : FAILURE in 30m 27s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/298f1355f2c842f0bed333e3b107e2f5 : FAILURE in 22m 34s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/777ac2acfe12481b9132f39c69738ab2 : FAILURE in 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8779e5fb62292f62010bf5487fc0d394a6f16851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 06:15:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ff22c5b4935f7add078a68431ab5bbba2c92b8e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 06:55:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/388e35c83e854e7fab7069a4dcd54dfe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c58923ad43b4bbda3058a5d991c19b3 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6818cd326e614ce4a064fd46cecbfb37 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69546043222949079f13866f8e0f36c9 : SUCCESS in 7m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97de31f45edc45d8b41831acebe94cc7 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4861f41205f4f799a8782969fccc61b : SUCCESS in 7m 24s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/836025481b0547f1a618f82a7e3d966e : FAILURE in 24m 21s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/151f28541d3f4f47b93d6cd871225ed4 : FAILURE in 27m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/345185b64adc45379565010b07a9a0f0 : FAILURE in 25m 43s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/94275a057c2547a78fdeeb1db131cb95 : FAILURE in 25m 45s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6c52c0bfe70f4f8ca213b927055c0b26 : FAILURE in 32m 14s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/8fc2033a4cc24c80a8a76cacca7615ca : FAILURE in 33m 14s","accounts_in_message":[],"_revision_number":2},{"id":"96328d5558137c0b47386193e2a725c27807ebc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 13:26:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d5d6ff6c58b2ac1c90b36ef2a6cd6f64745dec38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 14:39:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a66938a316a448da95c0cdfbefeaf4c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3b46e5641d43aab2d4f2756620fedb : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcf06f88cbe04331a6a63857b8099cfb : SUCCESS in 11m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ff75be17b644cf28621927314067b19 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b43cedfaf8940deb093c6a50877abc2 : SUCCESS in 9m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ebea22b768144cb89569391ebec2f799 : SUCCESS in 46m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d1c429a9b66a46fea36893e307d04340 : SUCCESS in 45m 12s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/660e07140e294a2da027ed8210d8e185 : SUCCESS in 51m 14s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ffebac9de8d04e2a84fba4e17b203356 : SUCCESS in 52m 41s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/4e771aef57044bf89c8f7fe26b53aa06 : SUCCESS in 58m 09s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/a4f424946f9341d7bab4eceaffe1810c : SUCCESS in 1h 04m 51s","accounts_in_message":[],"_revision_number":3},{"id":"003e762b4b724315e066671c67616d42fb6a1d65","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-12-15 07:26:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bdb4ecc7694b288e765b3c27bd876e80b74a61ba","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2024-01-09 05:24:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2a11c0d79ea9e422575b3503403158fdf0fb7e2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 05:25:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ff247ffe1a0d33a2b7df53f71873f0b3832fcacd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 06:46:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39613023606644a2879453f78ecf32fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0303498cf5cc4c4c827b7fb466efb495 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96e4596c58344fce88e2fdc68db0275a : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf516663affa4835aac840d4fcc0ba18 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb5c52902f34cccbedc0c8b67bc64d9 : SUCCESS in 8m 12s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6025f2c902a243239fd6e6d9a5d3821f : SUCCESS in 46m 12s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/c7660bf4f00a4774b8825f406c7bbf80 : SUCCESS in 43m 47s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d12fb4cf1f1040fbbd2548a922522bb2 : SUCCESS in 50m 29s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a47b4b0268f24fc0950b9a182f986c01 : SUCCESS in 40m 20s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/eaad3d55578a4e61abf25e726a7b9979 : SUCCESS in 40m 42s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/6ca6cb8f17ec4c458027f48d71bf518e : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":3},{"id":"0e670c778c056813a42c6163b6c18991373a61a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 06:46:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5c93c62d1b3140838922fa73e4bc5202d02f0fe3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 06:47:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccfb7a92ddd94a59bd063bbc47783e2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04fec00d56a45e7ada0767860fad160 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ac365555f46ef243716ff1824cebb42e84f5c305","revisions":{"393f21c65ea7cc50f05c3cf9b8575ee9e8f70d35":{"kind":"REWORK","_number":1,"created":"2023-08-16 05:19:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/891562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/62/891562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/62/891562/1"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 05:19:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 05:19:26.000000000","tz":540},"subject":"Dump loaded options during start up","message":"Dump loaded options during start up\n\n... so that operators can review effective values (except for secret\ninformation such as password) from the log.\n\nChange-Id: Ic094df537b5f7333be38c249cdec59a10d2e2660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/393f21c65ea7cc50f05c3cf9b8575ee9e8f70d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/393f21c65ea7cc50f05c3cf9b8575ee9e8f70d35"}]},"branch":"refs/heads/master"},"a14a15dbebb6a40a98117410893495f00f13f218":{"kind":"REWORK","_number":2,"created":"2023-08-16 06:15:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/891562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/62/891562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/62/891562/2"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 05:19:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 06:15:07.000000000","tz":540},"subject":"Dump loaded options during start up","message":"Dump loaded options during start up\n\n... so that operators can review effective values (except for secret\ninformation such as password) from the log.\n\nChange-Id: Ic094df537b5f7333be38c249cdec59a10d2e2660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a14a15dbebb6a40a98117410893495f00f13f218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a14a15dbebb6a40a98117410893495f00f13f218"}]},"branch":"refs/heads/master"},"ac365555f46ef243716ff1824cebb42e84f5c305":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-28 13:26:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/891562/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/62/891562/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/891562/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/62/891562/3"}}},"commit":{"parents":[{"commit":"38495d63cf25db3b04cd7c5743dfd609d9549e18","subject":"Merge \"tests: Remove duplicate fixture config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/38495d63cf25db3b04cd7c5743dfd609d9549e18"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 05:19:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 13:26:26.000000000","tz":0},"subject":"Dump loaded options during start up","message":"Dump loaded options during start up\n\n... so that operators can review effective values (except for secret\ninformation such as password) from the log.\n\nChange-Id: Ic094df537b5f7333be38c249cdec59a10d2e2660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ac365555f46ef243716ff1824cebb42e84f5c305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ac365555f46ef243716ff1824cebb42e84f5c305"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
