)]}'
{"id":"openstack%2Faodh~891576","triplet_id":"openstack%2Faodh~master~I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-27 12:11:58.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-27 12:11:58.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-27 12:11:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a","subject":"Use devstack helper for installation","status":"MERGED","created":"2023-08-16 09:42:14.000000000","updated":"2023-09-27 12:12:59.000000000","submitted":"2023-09-27 12:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"891576","meta_rev_id":"88b17fc5b50621c77f6d636b9101cbc4c23fd0cc","_number":891576,"virtual_id_number":891576,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-27 12:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2023-09-27 10:24:12.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2023-09-25 07:39:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2023-09-27 10:24:12.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-16 09:54:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-08-16 10:44:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-25 06:07:18.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-09-27 10:24:12.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"f14498386480007088dffd5cd94c003bfd8e7c55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 09:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efa519f10afeff43dd15f90914acb2762e80af54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 09:48:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8a77be63609aaa72dd2333b4fc35aa077ad3d63f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 09:49:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c0a2f9d8c6a02e3c9f2f47657c1f448a013a6271","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-16 09:54:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"41ec94f4a2025816c1eeee23fb4de5b497332fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 10:44:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e00e02a44294d23aec1ca0ea3b72710\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aab199efbf64fb9af689fb8cf1626d6 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f08cce27eec48e4bfba49419c3b88d4 : SUCCESS in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ef2eed7ffa2420f817b46e89fb79015 : SUCCESS in 9m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a02eb8cf8f6414f9fca66d6e464bc17 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ae509c092b414bb8dc7eb5a574e5d9 : SUCCESS in 7m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6cc3c88f50cf43f68ccb3d5dbba0ae66 : FAILURE in 24m 55s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/20541c4690e34911b4250e57e0c36d89 : FAILURE in 45m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/aa4244eec9e34a65921bb556be5d3f26 : FAILURE in 42m 57s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b037ff2d15df42639ae62c5a5020fb54 : FAILURE in 47m 25s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/99c75e2016ce4c3996b707284b3bb03f : FAILURE in 36m 45s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/89e4810657644b888ceaca882ab80267 : FAILURE in 45m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3909666360bea0de2f79eb3de6a06cb8ace81fcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 10:57:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7239605c49a52f3f40d5d608211b12460f082054","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 10:59:53.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"50aab6d1ba8d77ee82c8e2aa326f96d7734b9f26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 11:39:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/990bf7b1d337461fafb5e172133b78dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99a77cb487c348f7bb244567e6d161d9 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30cbcb67911c4580a65d942226a041de : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d5de1485d554e448f9e25b4cba89604 : FAILURE in 7m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19650b15c268459890bad977f9e8e9d3 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19218d5ad29142f78dfaebca558feda3 : SUCCESS in 8m 36s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/31b8d8c593574ad98902cc7870d9d40d : FAILURE in 13m 23s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/746ac933994f4942944fbc352203aed1 : FAILURE in 23m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2f1baec3dec64076a076b323da03a601 : FAILURE in 24m 11s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/78b8c5b63831479998d22e2975d2e0da : FAILURE in 24m 19s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/e619488dc0b846c3acbe5eb06a11b9f4 : FAILURE in 16m 57s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/be4b69aa7d4242f9bb3c099611a3f10b : FAILURE in 35m 02s","accounts_in_message":[],"_revision_number":4},{"id":"ee44642f544bfbf8a3a3cfc395ecd91319e8e4f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 13:04:50.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5f1d04254603223fcd6172eb367764c98f628e11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 13:05:06.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"491c8f9b9560a0af45fb95e09b35964685412f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 13:10:11.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"2fb3025741b5c911a5030e57d94bf2d2c94b1b90","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-16 13:29:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"461324e0efebe91a166476f1c2ddc30052eebb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 13:57:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fb38f9ec0154f90a42d2cf33a8d4234\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5182e4168e684194962440892fd5abfc : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5532fd03ca9b4b2a91b189e3aff4b20a : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f191f991a02641ddb2e93a65b6769a0b : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54ef4658fe6c49ff8b21a2c08fabb215 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5787ac1b7c4bb8b748fd8922587dc0 : SUCCESS in 7m 54s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a24029178a6444feb901102484c7c06d : FAILURE in 26m 54s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/022d1d9cd74a4102ba73b2f7a683b185 : FAILURE in 26m 51s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d43b028f68fe43c0b6a746babf92d512 : FAILURE in 39m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/78b9545afee044669b0ce644d1a95381 : FAILURE in 31m 04s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/674be007806f42b49c860bc6418c9693 : FAILURE in 35m 01s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/8fa49ce3858a44bc951a94e806898574 : FAILURE in 25m 16s","accounts_in_message":[],"_revision_number":7},{"id":"5229384e0ac5d329e1789a5ef97c299d1d3598ba","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-21 09:01:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"28c233bcca7e966569aa8f1801c4e72f4f9583e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 09:50:13.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/291a03c64ca245eab34fc68cdc88d772\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e20709eac7f542929619e5407959c0bb : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/872b7643b76a4e7aae52c1a86713c0b2 : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f8042eee980446098ee8b8d10fdd8f7 : SUCCESS in 8m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfbdb775e7ca47e393a4bd5fdf0c48b6 : SUCCESS in 8m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02d001e8085342358a436609c0753bfb : RETRY_LIMIT in 2m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83644c4bdac44d2a83bc4c2ef9f66e5 : SUCCESS in 11m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e25b949aa2e148aface81a187023245b : FAILURE in 35m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8faf8c3762094ccd820082ab96438573 : FAILURE in 43m 11s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a7900bb46426457fa002df9bbefa19f0 : FAILURE in 35m 04s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c8279a773157472b8ef43a3f947b70ab : FAILURE in 38m 18s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/2ff8ace0dd7c4800a89a03c9d794f13c : FAILURE in 33m 01s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/fa463bbc18de45509c2fc4c36f0b17b8 : FAILURE in 37m 54s","accounts_in_message":[],"_revision_number":7},{"id":"74ade481050edeb84bb440631e05e8b16eb1198f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-08-25 06:07:18.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1e1a0be6254e9a276d5f232f7aeab7158a497f9e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-08-25 06:07:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"665f7a1f288947720b819afad6709fc71dd49ded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 07:13:05.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8026df02f09446e4badaa6edc80a29ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab30c24545aa41c49031a07cfd594dee : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb1ca76d08fd41298f11a2cba260fb6c : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33de4562554a443389eaaccb42f996bb : SUCCESS in 7m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d6c5f2704be4c53822a675eafa8e114 : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0287ddd46ceb41d5803d2ad88764bd05 : RETRY_LIMIT in 1m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727bafc93ee94adc8ba16f17bc89fd03 : SUCCESS in 6m 37s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b4f40ba0715848f597f37b230ffed061 : FAILURE in 36m 49s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/548c7405266946278557bdec809fa3cf : FAILURE in 40m 45s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/901b1adef21d4b6b8c2bf9ff50d38440 : FAILURE in 37m 28s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/96894c680bcb4a95b8e854beaaf33979 : POST_FAILURE in 45m 11s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a2d9d88147b34d0d8a5204cbf7f7a9e7 : SUCCESS in 32m 16s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/700cc224654446349f357d01a8eeec4c : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":7},{"id":"933cf5b71d6376cc129b53e37855c4cf51d15e90","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-08-28 08:52:24.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f7ba8ac1e3ee0a5911571ab95823b4ffe89d6325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 13:14:13.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"279d03574336990af3014e7de2fa71320856d173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 14:19:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/715fe1088a31446e93a8a073a48478ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a0bea039c6466eb7fdcc7f4613dbbf : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd59a6594f954b45a84aed0c3372ae1d : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51f8064cffbb4084b529daccd71ccb4e : SUCCESS in 7m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cb73b26369c4901b72990ad0096b2b5 : SUCCESS in 8m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d5a5711324847cd803ec4a2db947509 : RETRY_LIMIT in 2m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c5f8c4076043a8972b9ff83919c506 : SUCCESS in 6m 58s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/34ea81a840a04c9690b1eb68efa73155 : SUCCESS in 34m 29s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a78f66e1af8a4fc3b8e769d23667cd53 : SUCCESS in 30m 57s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6d766a0938c44528a8fea90757007d48 : SUCCESS in 54m 10s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/036bd63969b14825848e22d9b9c0a59f : SUCCESS in 48m 58s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/61d89255aefc4a9daa84c59cc8fa167f : SUCCESS in 51m 12s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/81371cc895924ecdadc5ebca2cc86136 : FAILURE in 57m 08s","accounts_in_message":[],"_revision_number":8},{"id":"65d420f9c5ba0b8dfa1046bcc9504ca348894dd7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-09-25 07:39:49.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6a513202d7fd461c70e562e0b36904372e3926c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 08:50:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04db78c60ebd47ee91fae79b0957b5a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e429d4ea74b4ac9865fd952bacfecfa : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8db9e4704d8437c8fbd47c33c60173f : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3462eb4a00484241876859067a541469 : SUCCESS in 9m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e992280865b84378873f5c7a25cb920d : SUCCESS in 7m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0f14b4fa1dd47d2a00b921cfb231f68 : RETRY_LIMIT in 1m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c93cebdcf994cc8ad08c3da7165c704 : SUCCESS in 7m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2cf4b263236b44df838a78b1421ad0a3 : SUCCESS in 39m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7e957459e9d5420083b85f7e8d27b19f : SUCCESS in 42m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b1160c270f9f4335ad3866ee931dca91 : SUCCESS in 50m 34s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d1980cf3efdd4657ac3bdaac974eea33 : SUCCESS in 39m 31s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0de9ab332d2443e8bd24460ce7631bd0 : SUCCESS in 52m 55s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/5f7202d56a7d49cfa9349c293443aacb : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":8},{"id":"e72ea1c379ff36daa5ee441c5f0c0a00dbd0d32c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-09-27 10:24:12.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"368198b3e8ff374713b7a45e326a1048d5f4a155","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 10:24:51.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2e188e7055c3f6b94f7b50cf8f9fde22cf6e57ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 12:11:58.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33dbc1f3d15d483fa1f9de126167489f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985c87373c31438ebf5f31eb1b85c9eb : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77c491e115f340ec81944014de6c4e93 : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/072c6f21af8644e083211c4976358a70 : SUCCESS in 10m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bfd8d0f6e8742c0ada2bdd9bf717961 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba7d791a1e04d3f90e60317df23acf4 : SUCCESS in 8m 08s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/05ca87821cdd4a18961003905790de4e : SUCCESS in 45m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/688ec152be284a749a1b2f360910ab74 : SUCCESS in 48m 13s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4de08e6bb7de4b40b1b60293d40378a6 : SUCCESS in 52m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1e0639df10a74bc8941d66c84308e6de : SUCCESS in 43m 20s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/b58c92bc6d93464188dd4fb3a1cba159 : SUCCESS in 58m 26s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/a2fd23f7261346e3a40054755072ed76 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":8},{"id":"e6827935ed252b02d4f8dfb86c9ac1e7cd2c7cbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 12:11:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"88b17fc5b50621c77f6d636b9101cbc4c23fd0cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 12:12:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47e2d9d1d9654807836791daec6c2d2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1399369e3574ab3a574571e2a46ea5c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a55d9620e9d422dce1d2d9638ea0875bb58c7bf2","revisions":{"f1ee2aed1cf28b6814de7cb22babdf59813040fa":{"kind":"REWORK","_number":1,"created":"2023-08-16 09:42:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/891576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/76/891576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/76/891576/1"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"subject":"Use devstack helper for installation","message":"Use devstack helper for installation\n\n... to make sure the destination is selected according to some\nenvironment variables in devstack.\n\nChange-Id: I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f1ee2aed1cf28b6814de7cb22babdf59813040fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f1ee2aed1cf28b6814de7cb22babdf59813040fa"}]},"branch":"refs/heads/master"},"4f30444dcf5ca1fd956ec821a9851172e8df09e2":{"kind":"REWORK","_number":2,"created":"2023-08-16 09:48:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/891576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/76/891576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/76/891576/2"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:48:05.000000000","tz":0},"subject":"Use devstack helper for installation","message":"Use devstack helper for installation\n\n... to make sure the destination is selected according to some\nenvironment variables in devstack.\n\nChange-Id: I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4f30444dcf5ca1fd956ec821a9851172e8df09e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4f30444dcf5ca1fd956ec821a9851172e8df09e2"}]},"branch":"refs/heads/master"},"a5acfcee34d724127255a9323adb5f840188ede4":{"kind":"REWORK","_number":3,"created":"2023-08-16 09:49:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/891576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/76/891576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/76/891576/3"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:49:40.000000000","tz":0},"subject":"Use devstack helper for installation","message":"Use devstack helper for installation\n\n... to make sure the destination is selected according to some\nenvironment variables in devstack.\n\nChange-Id: I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a5acfcee34d724127255a9323adb5f840188ede4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a5acfcee34d724127255a9323adb5f840188ede4"}]},"branch":"refs/heads/master"},"c70c77652396347fffec428627b8cf803c96ef3c":{"kind":"REWORK","_number":4,"created":"2023-08-16 10:57:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/891576/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/76/891576/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/76/891576/4"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 10:57:12.000000000","tz":540},"subject":"Use devstack helper for installation","message":"Use devstack helper for installation\n\n... to make sure the destination is selected according to some\nenvironment variables in devstack.\n\nThis also fixes detection of the binary directory because\n\nChange-Id: I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c70c77652396347fffec428627b8cf803c96ef3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c70c77652396347fffec428627b8cf803c96ef3c"}]},"branch":"refs/heads/master"},"63f9bbd43b485e95d925aa23a6c36b0855dd37a5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-08-16 13:04:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/891576/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/76/891576/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/76/891576/5"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 13:04:50.000000000","tz":0},"subject":"Use devstack helper for installation","message":"Use devstack helper for installation\n\n... to make sure the destination is selected according to some\nenvironment variables in devstack.\n\nChange-Id: I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/63f9bbd43b485e95d925aa23a6c36b0855dd37a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/63f9bbd43b485e95d925aa23a6c36b0855dd37a5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"378fb840a8ef8169fce71adb2d0d1e951ecce1a6":{"kind":"REWORK","_number":6,"created":"2023-08-16 13:05:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/891576/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/76/891576/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/76/891576/6"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 13:05:06.000000000","tz":0},"subject":"Use devstack helper for installation","message":"Use devstack helper for installation\n\n... to make sure the destination is selected according to some\nenvironment variables in devstack.\n\nChange-Id: I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/378fb840a8ef8169fce71adb2d0d1e951ecce1a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/378fb840a8ef8169fce71adb2d0d1e951ecce1a6"}]},"branch":"refs/heads/master"},"5b339183c73fbedc5bdbeb06f970605507e62262":{"kind":"REWORK","_number":7,"created":"2023-08-16 13:10:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/891576/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/76/891576/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/76/891576/7"}}},"commit":{"parents":[{"commit":"94dc273712415910b951113633fe8a76ca9f72f8","subject":"Merge \"Add check-requirements job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94dc273712415910b951113633fe8a76ca9f72f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 13:10:10.000000000","tz":0},"subject":"Use devstack helper for installation","message":"Use devstack helper for installation\n\n... to make sure the destination is selected according to some\nenvironment variables in devstack.\n\nChange-Id: I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5b339183c73fbedc5bdbeb06f970605507e62262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5b339183c73fbedc5bdbeb06f970605507e62262"}]},"branch":"refs/heads/master"},"a55d9620e9d422dce1d2d9638ea0875bb58c7bf2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-09-22 13:14:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/891576/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/76/891576/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/76/891576/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/76/891576/8"}}},"commit":{"parents":[{"commit":"8de7585f374097bcb4910bfbed34b6c721490db9","subject":"Disable GLOBAL_VENV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8de7585f374097bcb4910bfbed34b6c721490db9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-16 09:41:14.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-22 13:14:13.000000000","tz":0},"subject":"Use devstack helper for installation","message":"Use devstack helper for installation\n\n... to make sure the destination is selected according to some\nenvironment variables in devstack.\n\nChange-Id: I1de8dbf693a7109b4fcceea98eab9f0782f7bd4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a55d9620e9d422dce1d2d9638ea0875bb58c7bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a55d9620e9d422dce1d2d9638ea0875bb58c7bf2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
