)]}'
{"id":"openstack%2Faodh~893986","triplet_id":"openstack%2Faodh~master~I29c476b4da2a0b0fda5c0d42df8470b801cc224c","project":"openstack/aodh","branch":"master","topic":"py311","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-10-02 20:39:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29c476b4da2a0b0fda5c0d42df8470b801cc224c","subject":"Fix bindep.txt for python 3.11 job(Debian Bookworm)","status":"MERGED","created":"2023-09-06 22:02:52.000000000","updated":"2023-10-02 20:41:01.000000000","submitted":"2023-10-02 20:39:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893986-py311","meta_rev_id":"1813f354a939b2c61be00389c5b8ee022056d066","_number":893986,"virtual_id_number":893986,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-02 20:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2023-10-02 13:59:56.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2023-09-11 08:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2023-10-02 13:59:56.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 22:51:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-11 08:52:50.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-09-14 09:04:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-09-22 11:47:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2023-09-25 10:25:22.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"state":"CC"},{"updated":"2023-09-26 09:20:26.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"CC"},{"updated":"2023-10-02 13:59:56.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"dfb657e883392229709f8210df4441d16a56381f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-09-06 22:02:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c93da0d8f47312e0a92e2337606dd301bb0b13e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 22:51:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1dd30e8bf0e4a5f90b2bb69b27c5947\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85b79b343174b969881b64b89b7b47b : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5214dd5e61f14bc9b7bd0f5e339b5f0d : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5fef99323ea490c9ddb15ae93b0efe0 : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab45cc2771824f67a50840e8824998bf : SUCCESS in 7m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93dc25d4e84f4d04bf6edb0efbbc5624 : FAILURE in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e0c724265341c8ad362cdf2a1069bd : FAILURE in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/731bef09bfe44d5da2581fbf6de9ebae : SUCCESS in 5m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b0cf78c7639c4339a961865f6b2c0cf0 : FAILURE in 24m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/86b220c549d84771b603ec376b83eeff : FAILURE in 32m 50s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1f7ac0ee364142d8891d687ec3d45e87 : FAILURE in 35m 11s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/47106b9e0af94dff9a19ac0b4933e969 : FAILURE in 40m 28s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a66aaabb2d0a4b818dc655d92dc0e0a1 : SUCCESS in 37m 31s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/01f893a469de426cb27372ab9c7fb146 : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a0903667f2b3bea7bce8e2c530441925833d35d7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-09-11 08:52:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bc8bbd54430d7e6ffdec7e1b5dd87dbec1fdd93","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-14 09:04:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0b699c1b700fc1ef3f8cf566f299c555f5fb9cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 11:47:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9f3280fb44386a7732a0e27d9de6c0930c22261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 12:51:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a7103434b8f44fb92fa5f690e269de0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ff45903f774b57a2fed6becd982853 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/debf509de7fc4bbda18d1602d6a898e8 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37aa42aaf50a43b196b7be73e84dcf06 : SUCCESS in 8m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a9dd2f153d84b8caca1915ff0a7e1fb : SUCCESS in 7m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84c3cbdd79e5478ca2bd6f8b58f97513 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4d3b0269524f8b8133bbfbae292b3c : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9977f59d407f4bb48fd7b7a1b3570016 : SUCCESS in 6m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a7d7a6946eab46328956744bdcb76b9d : FAILURE in 27m 52s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8336c4af07804056bd2e4dfebe528a76 : FAILURE in 27m 08s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c5f7a8dedfb14391a92e7eaaf25040b7 : FAILURE in 43m 11s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d58dbfebbc1648029dd66b4bad226885 : FAILURE in 36m 36s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/58a89fdd4e7245c590104321ece2a5ff : SUCCESS in 58m 53s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/9bb87367effc42c7a70e2330bca59b1d : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":1},{"id":"528cd883a97caf7a2c1d5a3dbc0e7dd2ff8f2e19","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-25 10:25:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"223654b4949ff5211ea4cfccfc9b69dde3ffc9aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 11:13:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06875520abce4577bfe8304f12ab84ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b63389f67e74389a8934760e6151961 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31ad9a6358dd4b9091c05056788ad4c7 : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55feb07176524ff1802b9f79be4f63a5 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcefeb433b2f429cb3808601fc079f0b : SUCCESS in 8m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0093386ca5054dd5a6450fdb902d3ec7 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754e0ef4e35b4c01b615e9512f6c3c6b : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20a0188d12624b75866f1aea22a611ff : SUCCESS in 5m 25s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9b2818ae839c4afe96ec97d980ebac75 : FAILURE in 34m 57s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f47ad946805043cc8c9e354dc784b709 : FAILURE in 38m 13s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/db9b50af7ee1418aab6a566fb2d8a787 : FAILURE in 39m 23s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/92dd5b8b05904fa4bec22bdf4a6a4396 : FAILURE in 30m 15s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/4a6a0c65a38e4a6c8838afe6385345eb : SUCCESS in 39m 03s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/2b37f42067244124b8a5b40ef5354560 : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":1},{"id":"bf1f095cbfb6d20e1dd8d73afc30083aae965426","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-09-26 09:20:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f87efd13c2bc89bb3d318e02874644bf3dd7168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:31:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9548a4049f374e6585f3b9150a5adc20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6e9bdd6fe24abd9ec02f79c251e0ce : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba725c5d5de24d3a9d662821aade2f2d : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3262cafb99a54746989f01384f09980d : SUCCESS in 8m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8225ec411334377be71045837bd4bd6 : SUCCESS in 8m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ec424bb104e44aeba48ad7513a9d10d : FAILURE in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c83eb4e5a14c61b750d94d8768415b : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/577a9babb28043fc85a543561c837978 : SUCCESS in 5m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/057005d3938a45308371b6980c368dcd : FAILURE in 42m 10s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/609ff88129ff4d98bc6c8bf03ee8aaa9 : SUCCESS in 47m 27s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9fed3f07c3b341319cc34ba4a69f53ab : SUCCESS in 48m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/748b8395d8b4410c97966d75e4de8ceb : SUCCESS in 36m 26s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/41c668b9c8a5410d960781e39d98565c : SUCCESS in 59m 35s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/95de8f0dc9a34780833fd0d38c781cd0 : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":1},{"id":"cf44619268c3ad1596c9ada13ce528b31d43f48f","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-10-02 08:35:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18d2c901cde0780dc04c0988907dc3f9dd89d768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 09:42:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5189eb1e1da64f63b74069cac7e28702\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e0ee9bc2d9439fb0733d53a343f0ca : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea7d2df16c6448b0b8d7229d83144e23 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/553bebf679884f6280b7bfc4e45dfbc5 : SUCCESS in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce42fa038b6a405bbd9d71503988d285 : SUCCESS in 7m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe2c0632fbc3425d91246bc2889dea89 : FAILURE in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d0baa237334dbb9b255a9b6f74c1e0 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a15903547be044a1a7b2c0d8fcdd921b : SUCCESS in 4m 44s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e24d74b63e95469e86d8a1ef7610b41f : SUCCESS in 28m 45s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/658f70f24837451985d3eb3cfc9fe443 : SUCCESS in 26m 51s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/db5bbfc1d7e74f6496e4cdba8eaf1c9b : SUCCESS in 29m 09s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/96d71dc8a12647408875d60ff6ec7b87 : SUCCESS in 35m 50s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/2ce9e88ede5f46d19e0f64c6b4244160 : SUCCESS in 42m 19s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/4ef4bf7c6436415cb0ec6a466efce40b : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e8698b164a762933fb2eba31ad1ae0720c4a416e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-10-02 13:59:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcabab71f29bf32086d0fa784b99f706912c0756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 14:00:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3bee990838326c24a35aacaeecf3e1f84be0b2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 14:58:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81ed766eddae4982b7806680d41fb8cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8b868fe8ba4927b990a59738f3116e : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97bcbdb5806b49019a175422ddde1ebe : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39f2da9f68b5448fad022b93bf2f6ed0 : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a750f6eb7514a16ab0b4b5d5b8c73c6 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff6e1aba2d72403cada44c91114fc7a6 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/974ef05c831f4d3ea1403a3c80ae0848 : SUCCESS in 5m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ce200a4fa72543ffa31ae5a5f8a3b268 : FAILURE in 27m 24s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f6e1b117401344e2a96656abaea6b8ad : SUCCESS in 25m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d142980341974ec38eb0e5799ef7f3e0 : SUCCESS in 49m 40s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e72c75c5d78946579b66c03d17208d5e : SUCCESS in 51m 25s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/3bb76b7e0f7949de9f808fb3a1c4b85c : SUCCESS in 57m 45s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/5e990023272542efa14123e5a075f308 : SUCCESS in 41m 52s","accounts_in_message":[],"_revision_number":1},{"id":"10282d7ad87cb8f689fc9554d61491eddf5d4a18","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-10-02 18:30:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79534e20704da22d2e5d0dbf7c33badc461544ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 19:38:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1da0069ea24c4dfeb4ce43f240f0d1b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcdfd976612f43a2a7f81069a871471d : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd436d29c6854f6db7350564a5897b71 : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/061c0b7918564f08b847c196c1a7a206 : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51c6f8a2f16b427ab221483bdaeb50aa : SUCCESS in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d683bd66172c4feebe6148d0dc7a3188 : FAILURE in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b71a82c669fa484999f60a9b8dd80657 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56556241c043481c9db185561aac7e76 : SUCCESS in 5m 56s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f604e6272be64b22a221b5470ef9d554 : SUCCESS in 43m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a381ea85cff44c9f8fd892bf5a32daa9 : SUCCESS in 44m 05s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/08b06fbfc66544229d8049ebf8a97393 : SUCCESS in 55m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6cd62e3c853641728dae94611695f681 : SUCCESS in 48m 42s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/c5650f0dc5954050bb03ed7a55ce7a81 : SUCCESS in 56m 05s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/7cc4e51ed44440c9b93ec389a57e094c : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3b1770e5d43b5d8f22395c41023f9260a81ab34a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 19:38:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5738292d91cb297894405c6ac29bbc3c7fae9e24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 20:39:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4bdc428d974a378d82166af8636ef7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889d51356a344a869fa514b5cd793e7d : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05bad2f0ecc74c24a52e6131017d9da8 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f1216270062492bae4fb725ec3ac8ef : SUCCESS in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d685379864ae4a6b90754d952be9d276 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40eb0d15b6ed49c5ac022ea6090ddc1f : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5157526e6c494b21b7894656c95544d7 : SUCCESS in 5m 39s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ff1b854fd7774bb3bfaa63f4a067d8ba : SUCCESS in 33m 36s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/028ea7a72aa343e6b0002d10d71e338d : SUCCESS in 33m 53s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cc0eb69e514c4c64a26b4957c7dd80e7 : SUCCESS in 27m 57s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/08f998889a07454b9d8a2f0dc36093ee : SUCCESS in 36m 48s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/434637037e0947be883e8460455adeec : SUCCESS in 44m 19s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/1ae32108add14df4b9fd056e428587a5 : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4ff249e27910c7414a726719a3b61e13a8708cfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 20:39:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1813f354a939b2c61be00389c5b8ee022056d066","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 20:41:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/021aee33f5cc4e04898b022c73327a17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cdd4442e3b648c5a6f54589669230bd : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46cf5a4cec714b34995f8a826ab43d58 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c3e3df3b2175fd2c1b96623596bdfc985a4a986","revisions":{"6c3e3df3b2175fd2c1b96623596bdfc985a4a986":{"kind":"REWORK","_number":1,"created":"2023-09-06 22:02:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/86/893986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/86/893986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/86/893986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/86/893986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/86/893986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/86/893986/1"}}},"commit":{"parents":[{"commit":"763bb401857e2501c92d7538a658df3aea38a07f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/763bb401857e2501c92d7538a658df3aea38a07f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-09-06 21:51:58.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-09-06 21:51:58.000000000","tz":-300},"subject":"Fix bindep.txt for python 3.11 job(Debian Bookworm)","message":"Fix bindep.txt for python 3.11 job(Debian Bookworm)\n\nPython 3.11 job run on Debian Bookworm which does\nnot the mysql pkg for example, mariadb-server, libmysqlclient-dev\ninstead they have mariadb. Updating bindep.txt file\nto adjust the pkgs as per Debian bookworm.\n\nChange-Id: I29c476b4da2a0b0fda5c0d42df8470b801cc224c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6c3e3df3b2175fd2c1b96623596bdfc985a4a986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6c3e3df3b2175fd2c1b96623596bdfc985a4a986"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
