)]}'
{"id":"openstack%2Faodh~894956","triplet_id":"openstack%2Faodh~master~I64aea43eddfa58e4d41d481736d97d06a089baa5","project":"openstack/aodh","branch":"master","topic":"release-health-check-cwr-bobcat","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-14 16:56:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I64aea43eddfa58e4d41d481736d97d06a089baa5","subject":"DNM: gate health test","status":"ABANDONED","created":"2023-09-13 19:16:17.000000000","updated":"2024-02-14 16:56:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1662e8bf0846b06563750b5de00c20508f1dc36","_number":894956,"virtual_id_number":894956,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-13 20:28:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 20:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb3bfcec6a8c1035af4bc008eebdc055cf9e807e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-13 19:16:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71700b3fb600f79a30437012c8c2c5c13d6ffa9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 20:28:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3697bb5f5e64f69b76c0b0df1d7d54b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d15ee3baed54ff8a7ef6f1f6b067521 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20947d34694b413fbae3d51a17cb5ff9 : SUCCESS in 8m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4645d11f15fc440c928552042ac38280 : SUCCESS in 8m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3551c9590564ecebab8f09384d2f167 : SUCCESS in 7m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d68ae6912ab44a58644227111c5eaa7 : RETRY_LIMIT in 2m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a85e29366942bfbae137b1808041e7 : FAILURE in 6m 15s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0150de0f69474097818aba0882319b65 : FAILURE in 35m 49s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f627387bcaa340e0b3526267bcccc349 : FAILURE in 38m 25s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cc50822b5f6a43cc9de7313d2afc5140 : FAILURE in 40m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/290dec9164c84ae1b2dfeb4a9336a7a0 : FAILURE in 23m 49s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/085e949afcbd4e7eaae430daad53bbc3 : SUCCESS in 1h 04m 19s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/5bf5b36ce3864444b9eae9f99ee93f6b : FAILURE in 35m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f1662e8bf0846b06563750b5de00c20508f1dc36","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-14 16:56:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f41fdc5cf08920e798ce5a7899f68260ca612b40","revisions":{"f41fdc5cf08920e798ce5a7899f68260ca612b40":{"kind":"REWORK","_number":1,"created":"2023-09-13 19:16:17.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/894956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/56/894956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/56/894956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/56/894956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/56/894956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/56/894956/1"}}},"commit":{"parents":[{"commit":"763bb401857e2501c92d7538a658df3aea38a07f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/763bb401857e2501c92d7538a658df3aea38a07f"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:16:07.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:16:15.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I64aea43eddfa58e4d41d481736d97d06a089baa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f41fdc5cf08920e798ce5a7899f68260ca612b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f41fdc5cf08920e798ce5a7899f68260ca612b40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
