)]}'
{"id":"openstack%2Faodh~895397","triplet_id":"openstack%2Faodh~stable%2F2023.2~I8c819490b47eef2223a79ef7d06f468686434de0","project":"openstack/aodh","branch":"stable/2023.2","topic":"create-2023.2","attention_set":{},"removed_from_attention_set":{"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2023-11-13 07:01:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c819490b47eef2223a79ef7d06f468686434de0","subject":"Update .gitreview for stable/2023.2","status":"MERGED","created":"2023-09-15 14:10:16.000000000","updated":"2023-11-13 07:02:30.000000000","submitted":"2023-11-13 07:01:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"895397-create-2023.2","meta_rev_id":"1d37ff34cee19b0b97ac79e75b32ac70284a1f0b","_number":895397,"virtual_id_number":895397,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-13 07:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2023-09-18 07:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2023-11-12 10:31:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 14:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-18 07:47:58.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-10-17 03:02:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-11-10 06:56:03.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"CC"}],"messages":[{"id":"a0afdcd330c1a2ea129482ac2926e6a8ee095eb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-15 14:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9650a6b0aa4f502dd6aeb4617b3d105a47b75eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:58:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d49ada48ce14d2e9fd0511de96c3d61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7bd51a50b24569ab3f8d104465acd0 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5a7ddaed9674d1b96d6b7834b0aff07 : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f4a561342df4eafae9d3b356dd74623 : SUCCESS in 8m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/318c233c3bac48dfaa86102f15c8bb8e : SUCCESS in 9m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c6a3c6ec42f431d8140f0f1ff99e5cf : RETRY_LIMIT in 2m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e75fb01b6445bc881dbaeee5b595f4 : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1833ed4babd2d856f5a323ff753d74331d929807","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-09-18 07:47:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81e3fa9f78534ded278c65dd383cbfff97030132","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-17 03:02:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f58e3c49f4c5290131e9363fc3adbc620a66d846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 03:34:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/644c2b5e7f3544ceae7f5238b4b87650\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82aa158a76314c9aa821e2edb2e0cf1d : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f3917a5fd5d43b3bb71562eb9861cf1 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b98a9da568b9470bafe1c60ee60a4788 : SUCCESS in 8m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/339c6986b2a84da6afcb175a0714a612 : SUCCESS in 8m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28159121ac1f4727bbb812a06add0a84 : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be77eaf68ac4b8aa15daa5730792a8f : SUCCESS in 7m 12s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ccef1d17387a4b79a0568f3ecce2f6fa : FAILURE in 25m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1379fcb297054a38b35288e4634d61b7 : FAILURE in 25m 01s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ef4e6e046be34e6eb4dd37ae59c1d580 : FAILURE in 29m 46s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/82cd867759f84414a53971d2de645280 : FAILURE in 26m 27s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/880041f94c5746d0957c1df302eda9c4 : FAILURE in 20m 40s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/bac87d37b72f4d2a9fae09c3be16595d : FAILURE in 20m 31s","accounts_in_message":[],"_revision_number":1},{"id":"464b94fea090728d4df82d5e1a0e55296e16b4b0","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-04 16:10:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c3f5f3c94d46e2c5a668ab50342dafc20578422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-04 17:18:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40bbcafba314417285b8dd6dfd76be57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3baf48fe21814d59a5ff0ce68b2853f4 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb4f7ed3b3e047908ad4e13fb6333ca2 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e32514e50c9c4191bd65d5c12cb06f23 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c990a9beb4747c4ba78381133984186 : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff166420b72d430c8fdee40da31cb841 : RETRY_LIMIT in 1m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de06856852a43c7b402d177f7b1d2d5 : SUCCESS in 6m 12s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b39ba2bf1b6740c7960ec5ad6ac984f2 : FAILURE in 18m 37s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/27fe731fbbe54a8cb24df9bcf941427b : FAILURE in 22m 47s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ec0b626472c04546911e0a7b7e83f316 : FAILURE in 40m 50s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/034b65cd7c2b4d9b8f16e07cd8eb3258 : FAILURE in 24m 45s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/75b4762ae3bb47aba8e21f8439112935 : SUCCESS in 1h 02m 57s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/2030fad47f3f4495b820ccbb0ee70667 : FAILURE in 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ec2048e494ea72c9a5accdbd85a26e0121bd7624","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-10 06:56:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfe62fb29d70592c3041ca78bdb738b2c5cac751","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 07:14:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90c4131ceb6ce975269d0eeced61c8232497dfa8","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-10 16:00:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b48a80184a58963f73d8ff119d2a82a654a247b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a6a4666d9cc434da058f22f9e991ed1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b5fd5211d0478dbd61670fb8fa5199 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66f1a70078364268be934a21aeba3350 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fad9e53ce724ebfb7022fb7947eea6b : SUCCESS in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dab2bfe8fce949378b6c671f62784192 : SUCCESS in 7m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9857ec624def4d52ab1d449dd7854d00 : RETRY_LIMIT in 2m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d6d4483c924d2cabc22a606b404d6b : SUCCESS in 7m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8ccb11b9bd2b49c9b77b3dd0377b65b4 : SUCCESS in 1h 18m 33s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7725396b5e0d48288292adcdb7070167 : SUCCESS in 29m 12s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/45f88a8146c84b34807d6aa04e1eda33 : SUCCESS in 37m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9563d49ec6094bf8b40d1e20aba409f9 : SUCCESS in 38m 37s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/64c0bf9d3a314fcab16ff3ec5f50b702 : SUCCESS in 40m 10s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/b3c17fbfede0407184e834de69bfdde7 : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6eba400344380284ec51993fda070ca115339d9c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-12 10:31:09.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13f9d4372e09136a435b40542cc8fdbef6fe5e35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 10:31:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c3fd7e13c6eb58a398d03600f8d7b6697059d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 11:26:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ba3735fe5954a71b3a521c35a9aa17f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ca39cd05fff436bab418961c7771881 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcb4bbdb296743b7896d678791aedb87 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/167a9a77a15149b9be781a2c470efcd0 : SUCCESS in 8m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7837a06a0f5f4b89875b8957c147c477 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85109e91fdf3479b97d79c892a10a7ce : SUCCESS in 8m 28s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f1f3774036384a5cbb3184b98157f8e4 : SUCCESS in 46m 50s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f2100f0f0b154a06946a0a30b4ed9cee : SUCCESS in 28m 12s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/5396c604cefd4e77959c63bbb2e7e652 : SUCCESS in 50m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ecebf340c12f4f50aec86824540ee620 : FAILURE in 48m 08s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/11c3f7acd1784634a0142547d0c3556e : SUCCESS in 33m 11s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/65e568f348b8497fbf291a1b317da772 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ac9e08de7091f7842162e28101a44f89f21312f8","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-12 16:46:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6e38c5ebebfbd334f05e450e751b67db279ab27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 17:48:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0527c4aad6364c18aa0c39b869d4b655\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e330677b75224dc190f36b571a82a928 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f586a6c7b2d4d52ad6b0bb37c0e7e7b : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b345b7551df04fa5814d42669b3ebee8 : SUCCESS in 8m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94493215d8fa4c2987dd318fdd354b90 : SUCCESS in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/730274f6fa7e4094bd2a9057e78cd242 : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88251474d6e4df4a4f74ef325438997 : SUCCESS in 7m 23s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6fb7fca34e944f19a615a4774cf4ff7c : SUCCESS in 48m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/943a913825054eac99b54bc185a22897 : SUCCESS in 26m 01s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/67fbae4909bd410fb3089c96b7c0bab2 : SUCCESS in 53m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/82aad1637d854b8cb13305d11a3fb0a6 : SUCCESS in 49m 38s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/16284f6a9ab2472787da2fffd99542cd : SUCCESS in 39m 36s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/5bb5d0f431594c97b56cd5056514c4a3 : FAILURE in 33m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3e9932c00883335a2dee38e0328bc84847729a6e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-12 19:09:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4a6f434bafed686f4fb13daeeaa5317cefb99fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 20:14:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/709ae295e03e4998b4a05e35b94fc823\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9133491bfd1241068e3233a8f1f240ff : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/994fc112893b4729afab1fc4f6407c9b : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87c7178ec70348b482b9e8440ce0607b : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a01812c6d8d46d88ab87a734ef26d30 : SUCCESS in 6m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eec01d4dd2744400bf3535453324e39e : RETRY_LIMIT in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee38a8c210b4b7f863e84248ca8a7d1 : SUCCESS in 7m 07s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/29d86052872e4cf1ab91fadcc99c8be0 : SUCCESS in 26m 54s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f08a3bf71cc24bd59ae9ac828cb6f944 : SUCCESS in 24m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0a50abf5cd434140bdb3d464ea367725 : SUCCESS in 56m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0323c646ecd64d8dab9689f466e18691 : SUCCESS in 26m 40s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6f445685399a424d98cd9a27d433ba20 : SUCCESS in 30m 01s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/bced66f8be2f43a5b913bbd2915a66bf : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a6f89adaca59240fe8ccb41131d5322f8e85be81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 20:15:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b34a5489aa0f725528030cd3a6e79713008eb87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 22:28:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75ca7cb1d24d454381096c9d20b50e95\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718b27e402774a078b078b56f30758ca : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc563d5fb03c476981cc95074c92cb11 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fa9d6df0bda438e92a6251efb956f4f : SUCCESS in 7m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8580a98d39844772958dd2040af5182a : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c61b38aae033459193e4aefa47e93ed0 : SUCCESS in 7m 01s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0f38f46fe9a64608834f6285b4daeeb6 : SUCCESS in 43m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/cbca187c2ee64dfc94ebe62d8d93bdb5 : SUCCESS in 24m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d4b8345a554e4f68a624e043faed0d85 : TIMED_OUT in 2h 13m 10s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/404b116e8d4d45deaf4642ac48ef9de3 : SUCCESS in 47m 12s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/364788d4ca6b405493204cbc2a2354a8 : SUCCESS in 30m 00s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c1374c6b4cb143a1988a1cf2dbd0bc3e : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9852b3cdb5726563cfcca9c4cfef0f4e57da4198","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 05:11:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5245c830b3af3cf3cdf357bb136bdb5728fd8a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 05:51:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ceecb92fa4243cb8bbe84989bc8fd4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a5837422854e1bb53b23de1a627a7e : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6772e44652ca40239f5617b67b93823e : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/992abf40180d4bb79789e0fe38dec903 : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc6d9c1517bb404b80b2b29c54358686 : SUCCESS in 7m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/823188c38f91401b996e1e6233a36f3b : RETRY_LIMIT in 1m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0df775521c4c89a6cf05da6a672726 : SUCCESS in 6m 12s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/487a89fd72a64767abd3bc87a92a698f : SUCCESS in 27m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f12864f2c5c247f18ef7d22ec5a81bc1 : SUCCESS in 26m 18s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1445c1889ae744ac85008ec4bfd6424c : SUCCESS in 38m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e6df7475e87c4168975e086da5aadb42 : SUCCESS in 26m 59s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/f0cd4a77f81445b18857dca910a94d69 : SUCCESS in 30m 46s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/3dfc2625dd594513a8d256ece5b463fa : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3cacfd7dbd79046421008ace50394522582a6685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 05:51:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"460e80bcef809b3d82b7c0a21274fc764f7139eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 07:01:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed938362cd7041d19d998e0410ecb471\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6102a9bcf6418a8109fad76e0f1691 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7568e07f70bc44fa916c2432f2b4535c : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d02443ec32c42058f0eea73523d3f68 : SUCCESS in 7m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ec96a86c076469b8a394f96e58d2b1a : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1bee39aecd4c20bd887a42c5f950b0 : SUCCESS in 7m 22s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d9aa7e27b8754273bf4e1718234c3886 : SUCCESS in 43m 16s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0fed594f49f146c996902408d6b785d0 : SUCCESS in 31m 42s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9a26f7989dc24cfd8715f4bb420a12d9 : SUCCESS in 47m 14s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/7b0ffc7403ca42e9abf5da8b5f0ce32d : SUCCESS in 27m 28s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/c5b0de4cbbb14582b06047e524cd3a36 : SUCCESS in 55m 05s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/7d057c75e486497fa7a5c804d79f3bc2 : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"80d9d58c3afdf2cc2201bc873bd4dfc5e1d422d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 07:01:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1d37ff34cee19b0b97ac79e75b32ac70284a1f0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 07:02:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2454efcc7c84391afd0b10be182e5d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd40664379745dc99c57bcd003f438e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27f1896814a590fcb939e3128a516a4d403665a5","revisions":{"27f1896814a590fcb939e3128a516a4d403665a5":{"kind":"REWORK","_number":1,"created":"2023-09-15 14:10:16.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/97/895397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/97/895397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/895397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/895397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/97/895397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/97/895397/1"}}},"commit":{"parents":[{"commit":"763bb401857e2501c92d7538a658df3aea38a07f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/763bb401857e2501c92d7538a658df3aea38a07f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:10:13.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:10:13.000000000","tz":0},"subject":"Update .gitreview for stable/2023.2","message":"Update .gitreview for stable/2023.2\n\nChange-Id: I8c819490b47eef2223a79ef7d06f468686434de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/27f1896814a590fcb939e3128a516a4d403665a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/27f1896814a590fcb939e3128a516a4d403665a5"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
