)]}'
{"id":"openstack%2Faodh~896087","triplet_id":"openstack%2Faodh~master~I539cdad868fe930a4cf660ed23290b05294292e0","project":"openstack/aodh","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-10-02 11:52:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I539cdad868fe930a4cf660ed23290b05294292e0","subject":"tox: Remove unnecessary configuration","status":"MERGED","created":"2023-09-21 13:18:49.000000000","updated":"2023-10-02 11:53:56.000000000","submitted":"2023-10-02 11:52:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896087-sqlalchemy-20","meta_rev_id":"528d26a8a633db89143c43ee434f70aa9fd09a61","_number":896087,"virtual_id_number":896087,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-02 11:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2023-09-28 10:20:16.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2023-09-25 06:03:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2023-09-25 06:46:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2023-09-28 10:20:16.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 13:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 06:03:14.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-09-25 06:46:48.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-09-28 10:20:16.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2023-09-28 13:48:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"967ff3ca97c9beea13452fa89a9522e97ab83ebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 13:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5e9a4da3ac4e56d5882f253acd658a4db2859b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:40:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e03afacf9a644eeab7ad048f22222d53\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87cf4b7a75843188e71a4707ee60125 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/604da04d7b84472e9b9116a2631f6f87 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9771d2b5e9b45d291d0959baa524116 : SUCCESS in 9m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/747a03b66fdb41c69f01669453d5e5c0 : SUCCESS in 8m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a63dce05d47d49d99ef82822e1083789 : RETRY_LIMIT in 2m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f53269dc214aa5b09a846b8b7a4502 : FAILURE in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88585b1321ad4736a77e06eea056b031 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8eec489549505a13fc79c048dfa7ecedeb9cf973","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 11:48:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b48405110592155f733924c782e0180c0edaecd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 12:05:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e3e6aea571447bba94765ff920aa3ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a1d32bc002484aa6e61f9f04c2b1e8 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31d71e8d105c4848a0988fc7385b0dda : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3db25c84ef44f40a22fe1485b4baf41 : SUCCESS in 8m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b056129e49a14ec9896a3bd9474446fd : SUCCESS in 6m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f86c661a2c5e48bab3d822a0bdd3ee3a : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf55405078e4006a09b577c56ab9555 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1981be2804584abe84ab2cf297b17de6 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d64757ab9e4a07a9c2ed8a0df975bc9d333b4d2f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-09-25 06:03:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6cc07839661bb28c9eb91506ad9504afe3d1415a","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-09-25 06:46:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c70c17cbde3ea93900eb7ed2e29a9da2771feaa6","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-09-28 10:20:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eef872bbe8d9ad2c7d4684041fe90a0dcf3c4d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 10:20:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40e75e1f160e332146bcadb1b72344de3738e72b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 10:34:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d49c340e35b246639c2d5aad170b11c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2287122b03d8464bbd5c65c2e0c4208c : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59a9bcb3e3744342918262d84bb14aa1 : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ef167734f4b4b24993be6fe31cbf9dc : SUCCESS in 7m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abc58be3688f477db0eb399f308018cb : POST_FAILURE in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee972147b5449129565f80553bd5ddb : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d77acf74e5e34c4e98408bbfd51776a2 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"49b32d49ba847364a510f7e0666aeefe5a27dee8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-28 13:48:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b396a446ed7b2394a311a0de25f7bde4180611b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:16:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cad3b545b5b24e6f8d24f0f14bf55949\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b0ecb6d0dac4e63acd3f2673e43b51b : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1244dcec43e2410ebec48f90ef2fe40e : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ee8369d5d07429d97d754ad7a70158e : FAILURE in 8m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d19a5f1c0d04444b5ab71ba5d31e19b : SUCCESS in 9m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0929a7a79e3944b6aee93b9fbbe7e705 : RETRY_LIMIT in 2m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b94eb52ec448299d8b87c7bfcb458a : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f922a71d616740e2803d27236c771757 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0214ce4a711c8e0483da30a2d0e4bb0d30765f5e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-10-02 11:25:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2194e8bc89aaa0785c7977636ca12407e72d94e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:40:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9660172c2d2549c3b3b8aa9ba640399e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e72137e945674eb9b7e637779bce20c2 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7faca0effcb54a9db3b687c002d1559e : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46a86c8aa0db4688b48b38a767d5b741 : SUCCESS in 8m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a66e59071e1410a9eadd53a084dfab0 : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d72e0d9fd2e4d79a108900f5d1af75a : RETRY_LIMIT in 1m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42989a9d7e104239a25e7b0925c1d620 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c064c844f2b9434a817f544b6024409a : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d1c32737eeb1592c7e7fe5ee0e819c7052de0228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:40:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"196ecca245e28e8b698eb720f9a45a27bca63904","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:52:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce5ba159d3a4fa2b2cdc5650e53de86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a487efe2ef7e46edbcf5f1288e874482 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26d51e3d0f5f42f189bdb20adf3be843 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab113190a6714ef7b7aceac54ad41c68 : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f93e46612f5242909181e9d1731f23be : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e33bd28718e48308ceb2740904d23ef : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/565530edf0cf44e29a011bab323c3edf : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3124e253f53f9c19e025bf53686a0d48381823b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:52:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"528d26a8a633db89143c43ee434f70aa9fd09a61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:53:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d5b49d129f24545b650dcc2ea0e969e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f684ea4a02a40a9a076c94f2e5aed68 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfd5fb9e20cc4fb79836fc31ef2931c7 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29b88f5db5785b8ab0ccc1de555a924fa8bd03f6","revisions":{"29b88f5db5785b8ab0ccc1de555a924fa8bd03f6":{"kind":"REWORK","_number":1,"created":"2023-09-21 13:18:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/896087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/87/896087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/87/896087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/87/896087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/87/896087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/87/896087/1"}}},"commit":{"parents":[{"commit":"763bb401857e2501c92d7538a658df3aea38a07f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/763bb401857e2501c92d7538a658df3aea38a07f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-21 11:52:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-21 11:53:04.000000000","tz":60},"subject":"tox: Remove unnecessary configuration","message":"tox: Remove unnecessary configuration\n\ntox 4 is Python 3-only. We don\u0027t need to configure a basepython now.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I539cdad868fe930a4cf660ed23290b05294292e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/29b88f5db5785b8ab0ccc1de555a924fa8bd03f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/29b88f5db5785b8ab0ccc1de555a924fa8bd03f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
