)]}'
{"id":"openstack%2Faodh~896126","triplet_id":"openstack%2Faodh~master~Ib81f3a59f8b1c836ee5944ca207a48b407480a8d","project":"openstack/aodh","branch":"master","topic":"fix-gate","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2025-09-05 14:25:36.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-09-05 14:25:36.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-05 14:25:36.000000000","reason":"Change was abandoned"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-01-27 13:01:05.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"hashtags":[],"change_id":"Ib81f3a59f8b1c836ee5944ca207a48b407480a8d","subject":"doc: Remove crud from config file","status":"ABANDONED","created":"2023-09-21 14:28:06.000000000","updated":"2025-09-05 14:25:36.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f04fb9e088f71c3a95f087cdf2aa2b324fe3c47a","_number":896126,"virtual_id_number":896126,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-23 16:02:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-10-24 02:09:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 14:41:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 07:40:14.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-10-02 13:56:57.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-10-24 02:09:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"817391ed5da6261c5e23786bcf31a506f2c7bd4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 14:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b71280a6d0a716035e9fe4e1cfb0fde8442f7f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 14:41:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10e4aeca749a476a92da8a6034300e4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e708bf8ff54c40a88ecbed24f013baf9 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d32099eddec4ebf89d04d59c43e6495 : FAILURE in 7m 45s","accounts_in_message":[],"_revision_number":1},{"id":"76e27238d251eec383b27bcdf683aa3b975f8667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 12:06:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"483768d3b7fac7bfbf6f8ed0dca011a1a0a27fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 12:15:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d78ccd4a1c36444a862a74e0561d77ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0184171d64d140bfb702ed5e8cbbebcb : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38475ab3dc194c359c65fb870f059c0c : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":2},{"id":"19e6e5888aa8a4f5d713eb1d482cc46792876f2a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-09-25 07:40:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"12fa6f1c33dae61a3c12546ed4dc3ba2309d8917","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-10-02 13:56:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9d9055d43ac162ad726fa8d6ae59d17b73a77b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 13:57:11.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/aodh for 896126,2","accounts_in_message":[],"_revision_number":2},{"id":"3f1edde4f7289e2e1319d18c176604e6880bb101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-16 14:51:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9685e89750ef9aa4343484ad43fe9dd5637236ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 15:04:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f32e327e1d8d4d5f92c3d1bf1a38770c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccbe38a1df84401483872907e33b2d6a : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c69eb9095cf34f14bf1cd906dd38847b : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":3},{"id":"aaeb64c4f94a9aded6a5306f66e8544c9eec3a0c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-17 13:22:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"54af2b2b6c87779f6c1c9be41cb16c03cfc62ce8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 15:45:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4cf0e27ddf13f727c316cc501ab2042f444ac152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 15:47:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"57c85a80ef5365b09c0ad1ee28adfe59a8a95701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:02:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6d682f9bd3c4c708f0e4ad5f23656f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f347d5f0ac4b4c69b56d4adb621d377c : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74734c1b34ab472fb3e48f51932dbfa2 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":5},{"id":"2c1f34062ade8d003230551ce65445f1f00d7ffb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 02:09:24.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9967b16268212cfea22a039b7edf42bb1f60ade1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-01-27 13:01:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f04fb9e088f71c3a95f087cdf2aa2b324fe3c47a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-05 14:25:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"39fea008974e849ff0cc87da116db63545e39cd2","revisions":{"303d3bb256937b4d288013a9adbd8acd922b16d6":{"kind":"REWORK","_number":1,"created":"2023-09-21 14:28:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/896126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/26/896126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/26/896126/1"}}},"commit":{"parents":[{"commit":"2a2958bd6338ed8b98fa165309fa8b44d83ac974","subject":"Fix docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2a2958bd6338ed8b98fa165309fa8b44d83ac974"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-21 14:23:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-21 14:23:52.000000000","tz":60},"subject":"doc: Remove crud from config file","message":"doc: Remove crud from config file\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib81f3a59f8b1c836ee5944ca207a48b407480a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/303d3bb256937b4d288013a9adbd8acd922b16d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/303d3bb256937b4d288013a9adbd8acd922b16d6"}]},"branch":"refs/heads/master"},"faf929918cfd37c6ca3d185b93ab91b4a95e894d":{"kind":"REWORK","_number":2,"created":"2023-09-22 12:06:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/896126/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/26/896126/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/26/896126/2"}}},"commit":{"parents":[{"commit":"2a2958bd6338ed8b98fa165309fa8b44d83ac974","subject":"Fix docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2a2958bd6338ed8b98fa165309fa8b44d83ac974"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-21 14:23:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-22 12:06:01.000000000","tz":60},"subject":"doc: Remove crud from config file","message":"doc: Remove crud from config file\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib81f3a59f8b1c836ee5944ca207a48b407480a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/faf929918cfd37c6ca3d185b93ab91b4a95e894d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/faf929918cfd37c6ca3d185b93ab91b4a95e894d"}]},"branch":"refs/heads/master"},"4e344564c2bdc491bae67ed432c122190f84c072":{"kind":"REWORK","_number":3,"created":"2023-11-16 14:51:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/896126/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/26/896126/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/26/896126/3"}}},"commit":{"parents":[{"commit":"197440c3c80ca601bd94719dadb6e6d338589349","subject":"Merge \"Add Prometheus evaluator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/197440c3c80ca601bd94719dadb6e6d338589349"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-21 14:23:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-16 14:51:05.000000000","tz":0},"subject":"doc: Remove crud from config file","message":"doc: Remove crud from config file\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib81f3a59f8b1c836ee5944ca207a48b407480a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4e344564c2bdc491bae67ed432c122190f84c072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4e344564c2bdc491bae67ed432c122190f84c072"}]},"branch":"refs/heads/master"},"2415b6d8777cf71f9aae3f56f9f7d9f07fe98628":{"kind":"REWORK","_number":4,"created":"2024-10-23 15:45:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/896126/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/26/896126/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/26/896126/4"}}},"commit":{"parents":[{"commit":"1980cc84d22f91e7bdd90f878822419bc4e1adf2","subject":"Merge \"Support file watcher to trigger GMR report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1980cc84d22f91e7bdd90f878822419bc4e1adf2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-21 14:23:52.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-23 15:45:16.000000000","tz":540},"subject":"doc: Remove crud from config file","message":"doc: Remove crud from config file\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib81f3a59f8b1c836ee5944ca207a48b407480a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2415b6d8777cf71f9aae3f56f9f7d9f07fe98628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2415b6d8777cf71f9aae3f56f9f7d9f07fe98628"}]},"branch":"refs/heads/master"},"39fea008974e849ff0cc87da116db63545e39cd2":{"kind":"REWORK","_number":5,"created":"2024-10-23 15:47:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/896126/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/26/896126/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/26/896126/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/26/896126/5"}}},"commit":{"parents":[{"commit":"1980cc84d22f91e7bdd90f878822419bc4e1adf2","subject":"Merge \"Support file watcher to trigger GMR report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1980cc84d22f91e7bdd90f878822419bc4e1adf2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-21 14:23:52.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-23 15:47:00.000000000","tz":540},"subject":"doc: Remove crud from config file","message":"doc: Remove crud from config file\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib81f3a59f8b1c836ee5944ca207a48b407480a8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/39fea008974e849ff0cc87da116db63545e39cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/39fea008974e849ff0cc87da116db63545e39cd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
