)]}'
{"id":"openstack%2Faodh~901793","triplet_id":"openstack%2Faodh~stable%2Fxena~I337dc6445a7ed43c386fedb0175c548688284e66","project":"openstack/aodh","branch":"stable/xena","topic":"dont test with prometheus before 2023.1","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-01-26 14:16:51.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-01-26 12:28:56.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2024-01-26 14:16:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I337dc6445a7ed43c386fedb0175c548688284e66","subject":"Exclude prometheus based tests from old branches","status":"MERGED","created":"2023-11-24 09:00:31.000000000","updated":"2024-01-26 14:18:51.000000000","submitted":"2024-01-26 14:16:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901793-dont test with prometheus before 2023.1","cherry_pick_of_change":901821,"cherry_pick_of_patch_set":2,"meta_rev_id":"a87db1ea31ad85a3196b3d9c8278309f918a6474","_number":901793,"virtual_id_number":901793,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-26 14:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2024-01-26 12:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2024-01-26 12:28:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-24 09:01:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 15:52:44.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2023-11-28 15:52:44.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"7eff0ddc6a78c280549ed00bca81374bd7f220e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 09:00:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"4a3d67b2933f550276e5a919ebfbd22d7180101b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 09:01:28.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/aodh on branch stable/xena.  The\nproblem was:\n\n  while parsing a block collection\n    in \"openstack/aodh/.zuul.yaml@stable/xena\", line 1, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/aodh/.zuul.yaml@stable/xena\", line 25, column 1","accounts_in_message":[],"_revision_number":1},{"id":"ecd1e16be3a318c4f190925b8a5221a0e4aee0a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 09:06:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0b9a16927053d8644d0aa292da9ef45457150f24","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 09:06:57.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"1adaa17b2ae52cb718c70505286a3627794ec34a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 10:24:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32a22e5c77d04ed89d85be8d7250c346\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a9d768db4148b2bdcc92f9d24e362a : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e267711968d427b8439876bf15dc443 : FAILURE in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da02edf5900e40178ea4cffae4e1ffab : FAILURE in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/544c246fc4d34dae8b7c300c3d4622fe : FAILURE in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db2be65ae62a4be792f779219df6e747 : SUCCESS in 6m 37s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/be0a18b509ec4aaabf85964fe325ca04 : SUCCESS in 52m 58s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/269dc000fc854523b7da958cd8304a7b : SUCCESS in 50m 16s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ba2eb72122ec42c382e5f67874ad4912 : SUCCESS in 51m 05s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/44bb202576714e12bea471abf0801dda : SUCCESS in 53m 26s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/d2cee9eec2d14df5943909ce728dd295 : SUCCESS in 1h 11m 49s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/82a92d76f3fe4fe99eb5867c01e9e3b6 : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4022c5b78acbd67c1c305e29d2f2f2db3d534915","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 11:02:31.000000000","message":"Topic set to dont test with prometheus before 2023.1","accounts_in_message":[],"_revision_number":2},{"id":"3a237c4318b6351fbdca1e7d6ee86424427d5790","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-01-08 17:57:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f0e1527540bd774bf1f8024814dc1d99aed09f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 19:17:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/134cbbc532c34e7dbb63a3b899be9ada\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04977fd1c4514922b877624e2978619e : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aac9c3cb99049b7b7e79eede225da2b : FAILURE in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c749abf9a7e4f398dfae2cc7e261d97 : FAILURE in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/764ffceb14ca44cabbe002573626bd6b : FAILURE in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572804ba78df4b3c9da30dde91a09976 : SUCCESS in 7m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b137255510fc4d6aa38cc7e35170caba : SUCCESS in 51m 52s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2e1a306c8f9a4ad58fb675788a1cee09 : SUCCESS in 37m 18s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/33794b5e1878467494958c129596d4ce : SUCCESS in 48m 36s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/dc4f7266acfe4bcabe8789f64bef4d2d : SUCCESS in 43m 29s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/c0727ecc6cdb4b7a903ef304ad852482 : SUCCESS in 42m 28s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/aab2d3700491459d907ef629dc10925a : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d4abab7d01a2fd014e27a421dd778cb424640222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-01-09 08:52:47.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0281183d77263e2b127891eae255f9b4c2b0cc4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 10:23:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c683fb75eb6c42c995a27b4e122a8678\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a415a21b744d429ca76ee62a7c3d556e : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c0bbcd3858a4f4fb4c1506018014d0f : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4158463482f4bc49f45628ca755f8b4 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/628561effcd6437f83ac171b851e0113 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56beba8e0464446a962b242ca1ae632 : SUCCESS in 8m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/32fa462543e548d199b38ab82440948c : SUCCESS in 43m 24s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0d7206211bb048b2a8ab365487ae4246 : SUCCESS in 45m 55s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2eea6cb0d5bb45379618adad5a5d9a87 : SUCCESS in 46m 15s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c383d9718e6640358713ad75f0747eb6 : SUCCESS in 48m 03s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/8ce3fbbfc69247e4afc6045b0d9711f7 : SUCCESS in 1h 07m 12s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/5fa420e4b67048dfb64334f9547668d5 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":3},{"id":"e509d7fd2a8fd1fe3e944881b8b664aefbf33a6c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-09 12:57:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"945895686f6ad177f15bd1568e0aba4f09b4f1ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-01-15 15:07:10.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"64a49fabf9b58f7a154b855f64962a6995cc09ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 16:26:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ccfef74826489fae27b33cb77fa8dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0e6b54f76642acb51f8bb1f5235f66 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df293256bb874ad1b3fd6b064d829645 : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7be68908bae34bfebcc3d8bdcdb80c8f : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/154f3f7494264b2aaa002bd8817f7572 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f833487b502e4765880c1a43a32df8b8 : SUCCESS in 6m 28s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/16c8ae566de04206b50e2ac9780cdb00 : SUCCESS in 42m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8cfa822927f246f48ced3b661ac7865f : SUCCESS in 51m 50s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d8bf67acead6472082cf3914526d3602 : SUCCESS in 53m 37s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a0be3ea027364caeab33d1681172d403 : SUCCESS in 53m 39s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/c6de773d3fe948949a88eb899ff71ca6 : SUCCESS in 1h 06m 42s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/ef62f259bfc94f6c92e406a1260af344 : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":4},{"id":"de6cf908a51511c040ba45ea2f33cdabb7099b82","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-26 12:28:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"076158003c60cde616ac6945539b0ba1a585c5de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 12:30:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c672be45c8121e6d88718939b6324467622e1139","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 14:16:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1610b98454348d69efbb92f25d1ebe8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5136d02d7c404ce591bcd43c0dc2705f : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/871c8e87fd4140c38208d6b5d9556f5b : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4b4ed9ae5704b0885a89d62c6d5b687 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd792e17052e47ce8273ce48ab933921 : SUCCESS in 5m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6bd17e47b7b0446aa0964a82cf59f12b : SUCCESS in 28m 47s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b081da584eab47488ca2d6cf98843990 : SUCCESS in 47m 58s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8f106ae8a5564887ad53b4044b83e516 : SUCCESS in 55m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0e4b6d69ef4f45f2931b3a934989b830 : SUCCESS in 54m 28s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/150c2eabc7c9454caa3b3460b3de4c3c : SUCCESS in 1h 12m 13s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/1e40555201854477ab4d4f80a87619c3 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":4},{"id":"18ff04611de37c6b5ab6550334e7faedd11e8ad8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 14:16:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a87db1ea31ad85a3196b3d9c8278309f918a6474","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 14:18:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/012035363cdd45bda85fe1454ee95d71\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0402aad00c1c4464a5e208affd7cfcfb : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d3df2295b50d3fb078e25d327f971b3793ecde0d","revisions":{"0652eac5de0cccfbafd5bb5f221900c49c46d63b":{"kind":"REWORK","_number":1,"created":"2023-11-24 09:00:31.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/93/901793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/93/901793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/93/901793/1"}}},"commit":{"parents":[{"commit":"e6d55b1ea9ee852cc56002c865c9443b526f2f58","subject":"gnocchi: Use Dynamic Aggregates API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e6d55b1ea9ee852cc56002c865c9443b526f2f58"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 08:57:16.000000000","tz":-300},"committer":{"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 09:00:31.000000000","tz":0},"subject":"Exclude prometheus based tests from old branches","message":"Exclude prometheus based tests from old branches\n\nDepends-On: Ic5bea08295d8c85303c676706415e093625ee16f\nChange-Id: I337dc6445a7ed43c386fedb0175c548688284e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0652eac5de0cccfbafd5bb5f221900c49c46d63b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0652eac5de0cccfbafd5bb5f221900c49c46d63b"}]},"branch":"refs/heads/stable/xena"},"66df3c89816f965636e6de8a389ddd6a61939e3c":{"kind":"REWORK","_number":2,"created":"2023-11-24 09:06:37.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/93/901793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/93/901793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/93/901793/2"}}},"commit":{"parents":[{"commit":"e6d55b1ea9ee852cc56002c865c9443b526f2f58","subject":"gnocchi: Use Dynamic Aggregates API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e6d55b1ea9ee852cc56002c865c9443b526f2f58"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 08:57:16.000000000","tz":-300},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 09:05:59.000000000","tz":-300},"subject":"Exclude prometheus based tests from old branches","message":"Exclude prometheus based tests from old branches\n\nDepends-On: Ic5bea08295d8c85303c676706415e093625ee16f\nChange-Id: I337dc6445a7ed43c386fedb0175c548688284e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/66df3c89816f965636e6de8a389ddd6a61939e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/66df3c89816f965636e6de8a389ddd6a61939e3c"}]},"branch":"refs/heads/stable/xena"},"833d6ce8c1b0ed2febe6eb35df9639200085882b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-01-09 08:52:47.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/93/901793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/93/901793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/93/901793/3"}}},"commit":{"parents":[{"commit":"e6d55b1ea9ee852cc56002c865c9443b526f2f58","subject":"gnocchi: Use Dynamic Aggregates API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e6d55b1ea9ee852cc56002c865c9443b526f2f58"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 08:57:16.000000000","tz":-300},"committer":{"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","date":"2024-01-09 08:52:47.000000000","tz":0},"subject":"Exclude prometheus based tests from old branches","message":"Exclude prometheus based tests from old branches\n\nDepends-On: Ic5bea08295d8c85303c676706415e093625ee16f\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/905071\nChange-Id: I337dc6445a7ed43c386fedb0175c548688284e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/833d6ce8c1b0ed2febe6eb35df9639200085882b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/833d6ce8c1b0ed2febe6eb35df9639200085882b"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"d3df2295b50d3fb078e25d327f971b3793ecde0d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-01-15 15:07:10.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/93/901793/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/93/901793/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/93/901793/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/93/901793/4"}}},"commit":{"parents":[{"commit":"e6d55b1ea9ee852cc56002c865c9443b526f2f58","subject":"gnocchi: Use Dynamic Aggregates API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e6d55b1ea9ee852cc56002c865c9443b526f2f58"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 08:57:16.000000000","tz":-300},"committer":{"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","date":"2024-01-15 15:07:10.000000000","tz":0},"subject":"Exclude prometheus based tests from old branches","message":"Exclude prometheus based tests from old branches\n\nDepends-On: Ic5bea08295d8c85303c676706415e093625ee16f\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/905071\nChange-Id: I337dc6445a7ed43c386fedb0175c548688284e66\n(cherry picked from commit 9724c77c236c103b5539795a24e35c5a64bc5993)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d3df2295b50d3fb078e25d327f971b3793ecde0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d3df2295b50d3fb078e25d327f971b3793ecde0d"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
