)]}'
{"id":"openstack%2Faodh~901794","triplet_id":"openstack%2Faodh~stable%2Fwallaby~I337dc6445a7ed43c386fedb0175c548688284e66","project":"openstack/aodh","branch":"stable/wallaby","topic":"dont test with prometheus before 2023.1","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-03-01 09:01:47.000000000","reason":"Change was abandoned"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-02-06 07:09:23.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2024-02-15 07:58:02.000000000","reason":"\u003cGERRIT_ACCOUNT_34975\u003e replied on the change","reason_account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}},"hashtags":[],"change_id":"I337dc6445a7ed43c386fedb0175c548688284e66","subject":"Exclude prometheus based tests from old branches","status":"ABANDONED","created":"2023-11-24 09:07:29.000000000","updated":"2024-03-01 09:01:47.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":901793,"cherry_pick_of_patch_set":2,"meta_rev_id":"97ba1a8d383f7c0314916c6f1864441f11ccbc43","_number":901794,"virtual_id_number":901794,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-06 10:53:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-02-06 07:09:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2024-02-06 07:09:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-24 10:14:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 15:52:33.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2023-11-28 15:52:33.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"5e2c7e988eeab555f2d5ece0a3ba0a9a1ec67b47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 09:07:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"7a81baa93eb5dfc3db87db1a149b9fff64c0e342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 10:14:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ddd78d8e7ff442e89ca28a62c08a3e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee9a96eb92347659852b3c806028dd9 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/696ae4561b94485bb3aa290d40c6d6ed : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a61d7c7fb0e74f0280e564fb7448ded3 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/297b75e92fc6471bb9cf38c891424f55 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5b3ec909784e6fa8594da508fba0d7 : SUCCESS in 6m 46s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/407f4d858e604167983ac54dd1e6f764 : SUCCESS in 48m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/42a4b9fc2de745b592a36de0b8f78a8e : SUCCESS in 36m 44s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/00972a26cf2a421794a607d2a80fa275 : SUCCESS in 59m 23s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b99865d2667a4f0389688c99b6a1214f : SUCCESS in 59m 54s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/461cd52c1e664169811fb501174f08c7 : SUCCESS in 1h 00m 45s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/d3f70fa73a6b4e149ddc3418e93c6ad0 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e6b4ebbf2310cbf90a954ab96b2905225cf8dea2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 11:04:05.000000000","message":"Topic set to dont test with prometheus before 2023.1","accounts_in_message":[],"_revision_number":1},{"id":"f196ca4785a90c164250a5d40e5f91721c1fdc18","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-12-20 14:55:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c8905f28ed6075cb14399e57a7b58d7aab551436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-01-15 15:07:35.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3324067b4b9d25ebd30cba563bd3fb841408a6b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 16:46:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094d38b2eb4143e0814e5e53a7fe58b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fceb87ec6e3f4208bfc59090a5d4225f : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd5069eb38d64ea382a89ceb1151ea59 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78d7b5791c8141fda313ab1f26155c50 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de04faffe7cf4a9e83ef271fcc8143b6 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43cb925ffd49441f965013abfba574a3 : SUCCESS in 7m 25s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ba39df41ce50496db07d274fe2dbc555 : SUCCESS in 56m 50s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1a45d50a7c6f47bdbc55298b3800cf03 : SUCCESS in 52m 53s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d4f05aabd2764c9b810d7edc316a797e : SUCCESS in 1h 04m 27s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/56d213598c4341a9bc40903b367c05a4 : SUCCESS in 39m 40s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/540039192b3847be8967ecf86c3cb927 : SUCCESS in 1h 12m 38s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/5852b51c0d9a432c80eddf1e4c36147e : SUCCESS in 1h 31m 10s","accounts_in_message":[],"_revision_number":2},{"id":"81a82bbf49fa3246bda372f67c33b932d0b2da06","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-02-06 07:09:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"adfb60f96ca5c30783fe202f4c9e2e8e99fa838b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 07:10:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1a9cccbf459af34f07ab3db6b1a36b176480cd57","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-02-06 07:10:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"751484a81f440976ba519e5b1b89b627a58647d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 08:31:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eea77aefd87c4b9b803ddae120d35b94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d6065f13e341718603494545ea51cd : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2c10e9f5cd340d4b854f12051f62aa7 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/044421b761bb42a691312a962f98f571 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef1adea0eb047338e5823448877cca7 : SUCCESS in 7m 39s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/1b9c034cbd9f4975b54f2f27a1ca1130 : SUCCESS in 42m 15s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e9a4d87b5ba5466cb062a00eaf7a5839 : SUCCESS in 38m 36s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8edfb98225c64dcdb410a3a14f757cc5 : SUCCESS in 49m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/96417e78f8d74eaab3e29d27db5ace0b : SUCCESS in 48m 03s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/5ffb76c5028a4b4980dff5436704aeac : FAILURE in 1h 11m 24s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/314f94c28f3b4518847119b26d5a559d : FAILURE in 1h 14m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f3572625f6a562a29123f0bc5f4b9c5a6e100b7a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-02-06 08:36:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"864346ea8d0fa0ad040d65591a414c2b4f792005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 10:53:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebbcd62c438d444b954486d9f09a279b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f025d0c424704bf1a62b9dbc70ded8cb : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a348a4ea504efaa1bbc5169bc0c74c : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4caf86ccdb7f48d6a2780aa65c7bcfad : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86693004f54248069df7073714434b0e : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0491f5209af24934a2fd795575c62812 : SUCCESS in 6m 16s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6b38dddc168a4747881500107646658f : SUCCESS in 59m 57s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f5926d2c640e4c3887fde4f5633944c8 : SUCCESS in 55m 01s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3fc0ee0ea2e645e590318945bd31a6d9 : SUCCESS in 59m 11s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/064c5d541a704f1a96ce339a7ee00c96 : SUCCESS in 58m 55s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/ab61d4b32cf34a80ac2a880be478ea28 : FAILURE in 1h 22m 37s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/263733de504d4cb6a9377c418cd865ee : FAILURE in 1h 50m 07s","accounts_in_message":[],"_revision_number":2},{"id":"58751e349af7123e47c5fdf607e991f70b900ddc","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-02-14 07:03:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5065eb0dfe4622e20e5775f5b524c907cdd7b4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 08:25:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fe2a11444e043de84c1e85d07d573d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd21a80fbff24d30832f913a8ea0d7ef : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d49938a410b45729ebd873cb6ac4c2d : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5783cd24138e472386b8f41e3e543d23 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6d6050442814cfaa4184a431847eba8 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0c4971fbdf4edb90149e3459a55a4c : SUCCESS in 6m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6ff3042eb2f04aca86f65be1a29c7864 : SUCCESS in 57m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/57b00bc4ce9a4e5990548e06f3ca02d6 : FAILURE in 39m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f1908c87b78b45e991df699cffc25e68 : SUCCESS in 1h 04m 39s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c389eac403b54472967544a252818713 : SUCCESS in 1h 02m 13s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/a6655338263e42bbbb48d34b544d2301 : SUCCESS in 1h 02m 26s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/7de521a45ac841b58d6886d77f386ceb : FAILURE in 1h 19m 59s","accounts_in_message":[],"_revision_number":2},{"id":"935ced6dbeeeecc57d78f37f7a9e7954b691e052","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-02-14 09:56:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa2c1b275227f403f4528d583493f854dbb1635d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 11:40:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65d15e59ea8e4601b800fa38c200e31d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c67fe011dd469082184cbcad7b1b1f : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/956d37fa0ab74b1383ab0733ebd6b73b : SUCCESS in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5113ed9dae3f4eeb9d97feb18470b355 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b6e03aa49254816a00284c4e82d3294 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/265689b2eaf44be6ac89887eb0d702f0 : SUCCESS in 6m 30s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/272c009729b44095b4a32371317bcb87 : SUCCESS in 56m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/26ca12576663496da88e57fc0e407172 : FAILURE in 55m 09s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/28a196798b74449696c6039625f9b8bf : SUCCESS in 57m 45s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ad09aa98908a414dbe51775dbf07aac9 : SUCCESS in 52m 18s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/f880227eacfd437b8f1cccbcfc66d4b4 : SUCCESS in 1h 30m 02s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/9bcc3ee66eb349b68ed18fdd9aebe814 : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9ef10748ae90033184b5ddb6f5df49a95c4919f2","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-02-14 11:50:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e330b867c4f712d88f5ecc5266d9764e038e2d7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 13:17:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cd83a3821c74ff9b099f2f344b2b84d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0270bed659648759e98d6e8934f7715 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/031db0a61842467e8c4ba7438d987540 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d29ef62bf7fd428ba812059a80d134a9 : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/613ecca9253741778bebf5b2de0f5b7a : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73cbe0f0eea4b4fb6b3eac36b19d232 : SUCCESS in 6m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/476e2ebeceeb4281aab7a2327137a92d : SUCCESS in 58m 46s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/eeb3500f1c2148cfb39380017a384399 : FAILURE in 44m 10s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/27dcbc2663464579b72654af1afc9fa2 : SUCCESS in 50m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f545538877364b0d8ca961efb15d2d62 : SUCCESS in 45m 19s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/5413829dedc547e1bda2b1615efc505a : SUCCESS in 1h 03m 39s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/f44b10bf22734ed19ed4d83923d3a3fd : SUCCESS in 1h 23m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4191d47dc675b9b02ecfba8cb479f1e7225243c7","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-02-15 07:58:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97ba1a8d383f7c0314916c6f1864441f11ccbc43","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-03-01 09:01:47.000000000","message":"Abandoned\n\nAbandoned as part of transition from stable to unmaintained. Please repropose to unmaintained branch after creation if still valid interest.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"75766988eb06dae9d8e8b0cf55284984ce439bac","revisions":{"379d0ffa419ce2976f8350f093c3b2eed6609b20":{"kind":"REWORK","_number":1,"created":"2023-11-24 09:07:29.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/94/901794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/94/901794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/901794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/901794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/901794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/94/901794/1"}}},"commit":{"parents":[{"commit":"9ac090bb37cb628c6867143312e84b57346bb97e","subject":"Merge \"Remove TripleO job\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9ac090bb37cb628c6867143312e84b57346bb97e"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 08:57:16.000000000","tz":-300},"committer":{"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 09:07:29.000000000","tz":0},"subject":"Exclude prometheus based tests from old branches","message":"Exclude prometheus based tests from old branches\n\nDepends-On: Ic5bea08295d8c85303c676706415e093625ee16f\nChange-Id: I337dc6445a7ed43c386fedb0175c548688284e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/379d0ffa419ce2976f8350f093c3b2eed6609b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/379d0ffa419ce2976f8350f093c3b2eed6609b20"}]},"branch":"refs/heads/stable/wallaby"},"75766988eb06dae9d8e8b0cf55284984ce439bac":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-15 15:07:35.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/94/901794/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/94/901794/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/901794/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/901794/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/901794/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/94/901794/2"}}},"commit":{"parents":[{"commit":"9ac090bb37cb628c6867143312e84b57346bb97e","subject":"Merge \"Remove TripleO job\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9ac090bb37cb628c6867143312e84b57346bb97e"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 08:57:16.000000000","tz":-300},"committer":{"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","date":"2024-01-15 15:07:35.000000000","tz":0},"subject":"Exclude prometheus based tests from old branches","message":"Exclude prometheus based tests from old branches\n\nDepends-On: Ic5bea08295d8c85303c676706415e093625ee16f\nChange-Id: I337dc6445a7ed43c386fedb0175c548688284e66\n(cherry picked from commit d3df2295b50d3fb078e25d327f971b3793ecde0d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/75766988eb06dae9d8e8b0cf55284984ce439bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/75766988eb06dae9d8e8b0cf55284984ce439bac"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
