)]}'
{"id":"openstack%2Faodh~901795","triplet_id":"openstack%2Faodh~stable%2Fzed~I337dc6445a7ed43c386fedb0175c548688284e66","project":"openstack/aodh","branch":"stable/zed","topic":"dont test with prometheus before 2023.1","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-11-30 09:16:24.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2023-11-29 11:09:33.000000000","reason":"removed on reply"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2023-11-30 10:39:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I337dc6445a7ed43c386fedb0175c548688284e66","subject":"Exclude prometheus based tests from old branches","status":"MERGED","created":"2023-11-24 10:39:23.000000000","updated":"2023-11-30 10:44:51.000000000","submitted":"2023-11-30 10:39:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901795-dont test with prometheus before 2023.1","cherry_pick_of_change":901821,"cherry_pick_of_patch_set":2,"meta_rev_id":"e8798b47bb3384f4e3799772006d3608d7e1a9e3","_number":901795,"virtual_id_number":901795,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 10:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2023-11-30 09:16:24.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2023-11-29 11:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2023-11-30 09:16:24.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-24 11:40:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 15:51:42.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-11-28 15:51:42.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"49d75a786a9354d48cb2a3b93f36e50b78b0ef6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 10:39:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"6f60204802c9404e262065ee90976c49686b2c07","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 11:03:45.000000000","message":"Topic set to dont test with prometheus before 2023.1","accounts_in_message":[],"_revision_number":1},{"id":"9b2b67a73752a6eae2d335bacab09ba528f99a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 11:40:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14b4369bab074e808ac830d80d613cc7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff8ef650b94494f95a873b7f7d4c69b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15975cb2ea284236bfdaa289f4403f4e : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb5c7bec075847e59b24c06f37a6700b : SUCCESS in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79e1ea68b7274f55908ab89d7d12f5cb : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07edbb2573ab491e9599a905c0b2fdbb : SUCCESS in 6m 55s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/be7a5b6dc08c4beb88f86b877bf6d69c : SUCCESS in 42m 03s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/52bba078193c4cfc8e2b119363a36c5b : SUCCESS in 44m 21s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f55078ba91df48cb93ada1e86329f37b : SUCCESS in 52m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/31ce32c6b98043f4a72d17d6d8bba1ee : SUCCESS in 32m 56s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/12d20878b8614dc5a0537da51d06d8da : SUCCESS in 42m 10s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/8e9dc464ee864126ad88898f5e2e0156 : FAILURE in 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5960b052cbdb3632185bc50a3f40b9080c911c1c","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-24 13:38:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cda54700d2c96ab8f5c2ff8e98fa0ca65c860647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 14:55:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a8640e10cc40f9a68d96b29e45823a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca594726b384a7aa46e5ce0df5e5c51 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dc4b457e5bd4e6ba8b2d42f33445dde : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a788de73ac6b4d4abd211780807173d0 : SUCCESS in 7m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1787d913cf4f490395c49b9539adb010 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f284926961c94db48d10affba010bc91 : SUCCESS in 7m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/06f37907b3354cf8913c00e8e1f78e6a : SUCCESS in 51m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f6de2356179840d2a7b956796a058ca8 : SUCCESS in 46m 00s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/729ff35a130446c0a852ba80be77d372 : SUCCESS in 38m 34s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3afd8fa7fda4499ca80ff62fcecc8353 : SUCCESS in 45m 41s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/b9257363ed9347c3a14e0531039071ca : SUCCESS in 1h 02m 31s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/886bc31e7b6645aabba5d966bae51f2e : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"39209dc0221fb0be619a38f252a52502d808bc89","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-29 11:09:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"138a8084373d6df450e8e9d2ab233d21129a9475","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-11-30 09:16:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"280df3d5fae38b74a3684188f0bedbbc3f62cc13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 09:27:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d23a9d22b17de3a655d489e950a05e2c41ab0fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 10:39:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/120521333b8d467199553f0689f6b96c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d47ed3bdf364ccc8f42e341e0de8162 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/924dd1a7d7104d9182297cd4303978d9 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b49171d600f4ae7a4e13ba987326ec9 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db737a494421491abead33150154a8e5 : SUCCESS in 8m 29s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0e972b9f1030415892c9f06b39b1d549 : SUCCESS in 47m 42s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/afc5e6e346334bf9bfe697078924b16b : SUCCESS in 54m 00s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a83b5fcf395b4df98c29bfdb7e7fd3d2 : SUCCESS in 49m 21s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/db58fdf5969243e3bb264927e3a69d00 : SUCCESS in 35m 15s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/97a80254ec4d48b3ad7f2de3edf23a71 : SUCCESS in 52m 05s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c8d6939991d04cc6af6ee6daa070b26a : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a8f27b98b666560daa7bb3315ce548208fab2df0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 10:39:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8798b47bb3384f4e3799772006d3608d7e1a9e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 10:44:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b1ee9a772dd49228990080ca6a2ce4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51255a8ab1344835b8b98a19f8c3c4bc : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d4fb04e36610d208e5ab6a32b25ebfbd1a2e590","revisions":{"6d4fb04e36610d208e5ab6a32b25ebfbd1a2e590":{"kind":"REWORK","_number":1,"created":"2023-11-24 10:39:23.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/95/901795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/901795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/901795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/901795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/901795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/901795/1"}}},"commit":{"parents":[{"commit":"27b3cd2fcfcbfd9265cdbbfd3e7cd391166aae82","subject":"Complete the revert of \"Support Ceilometer API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/27b3cd2fcfcbfd9265cdbbfd3e7cd391166aae82"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 08:57:16.000000000","tz":-300},"committer":{"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-24 10:39:23.000000000","tz":0},"subject":"Exclude prometheus based tests from old branches","message":"Exclude prometheus based tests from old branches\n\nDepends-On: Ic5bea08295d8c85303c676706415e093625ee16f\nChange-Id: I337dc6445a7ed43c386fedb0175c548688284e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6d4fb04e36610d208e5ab6a32b25ebfbd1a2e590"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6d4fb04e36610d208e5ab6a32b25ebfbd1a2e590"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
