)]}'
{"id":"openstack%2Faodh~902374","triplet_id":"openstack%2Faodh~master~I5f1fe1a02dbccfca48c95e97657c72054016a17e","project":"openstack/aodh","branch":"master","topic":"separate_backends","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2024-06-04 14:05:27.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-06-04 14:05:27.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-06-04 12:52:55.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2024-04-17 16:16:07.000000000","reason":"\u003cGERRIT_ACCOUNT_13177\u003e replied on the change","reason_account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}}},"hashtags":[],"change_id":"I5f1fe1a02dbccfca48c95e97657c72054016a17e","subject":"Add tempest configs to jobs","status":"MERGED","created":"2023-12-01 11:49:11.000000000","updated":"2024-06-04 14:07:41.000000000","submitted":"2024-06-04 14:05:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"902374-separate_backends","meta_rev_id":"69894a3bc2d6e06230f7347bbfed31837f21221a","_number":902374,"virtual_id_number":902374,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-04 14:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2024-06-04 12:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2024-05-29 12:40:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2024-06-04 12:52:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-01 12:59:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 12:17:16.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-02-12 11:53:13.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"CC"},{"updated":"2024-03-07 15:50:58.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2024-04-17 16:38:10.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"}],"messages":[{"id":"d5e78876ac10e19e31c4f688a13bf0ebdd2f7261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-12-01 11:49:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f9e2369fb2b0f5383ccddd68b8c28e9ea7e092e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 12:59:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a18a916cbbc499490c4bb5e4fe732cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353e86249ace454998784907a30319ee : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aca461d34f4e42bfbccb8773943b30de : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27b98e3ecde341cf8c59c650a1819d11 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210f84860c6d4fdb9d9f151da74674b4 : SUCCESS in 6m 29s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6a18bfd2b4da4935b170616699057bf9 : FAILURE in 29m 07s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a829129c91e041b49477bcd1de328674 : FAILURE in 31m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a175b26226e44434b631cffb7f662219 : FAILURE in 47m 22s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/7530ae58dfc04e7da1c28d962b6bfd7f : FAILURE in 39m 53s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/2b443fbca1204cad96c3fd57ac0380f0 : FAILURE in 53m 26s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/a30107badcbb40bd81f47b826c60364e : FAILURE in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bf34541b3d2e949690cbc8df5a7f86726c800198","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-12-04 10:42:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"570df17d7cd9697c31253da63eabcc7e687c0dbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 11:48:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a084b3a90e81448e9bcee2a60d074b25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3997e65f58b4c2c9ead309e24cf9d57 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8bebe1f4c9442369eea40b99772b11d : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b9b7dcf4ff24464baa90692887679c8 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057b9d0f9fd0419890b0c987bc916787 : SUCCESS in 7m 50s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9190e2a4a81a49d5a7c236fa54c5bbbd : SUCCESS in 56m 30s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/979aabfee34145de9ea69992fc7ee06b : FAILURE in 53m 01s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/29f33bc525854bcd9c9161874132a8e6 : FAILURE in 34m 25s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8979fa58e01e461d8315cd18f3a9253e : SUCCESS in 49m 19s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/27440680f8ac4592be083323282a68bd : SUCCESS in 58m 05s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c79a3c880fc7405d916148e738e456d5 : SUCCESS in 45m 56s","accounts_in_message":[],"_revision_number":1},{"id":"56b77679fcb0bf8cd9bad8a959eaef743dc877e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-12-06 10:08:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"752066195681aca2d0d5dfb32644c84b26eb6f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 10:09:29.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/aodh on branch master.  The\nproblem was:\n\n  while parsing a block mapping\n    in \"openstack/aodh/.zuul.yaml@master\", line 44, column 13\n  did not find expected key\n    in \"openstack/aodh/.zuul.yaml@master\", line 45, column 15","accounts_in_message":[],"_revision_number":2},{"id":"70280923bac5076c12e8a32f329cf74e6292432d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-12-06 10:20:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"efa80f02e3616df6a4297513ec3a1019740f546a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 11:22:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d0419d8f8e44516908b5fce7baf3249\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbeb0f5bb964444d8b90f0d79a785b4b : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48181e2450ff430bbf97621bbb36a95e : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4697194f04d64538825b470d19e44ecb : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd56f27972264d17a37cecee0d347fb6 : SUCCESS in 7m 22s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/1093bb50d9cf4d9186c8341e5696f46e : FAILURE in 46m 17s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/28cd231a1a3f4332a487c133b17d48d3 : SUCCESS in 30m 34s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/096ebeeb19584fd1becf84ddc5bb220f : FAILURE in 53m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f655873bf69242a99dfa75d9a8db71c8 : FAILURE in 38m 36s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/1fa00cbda05a46409a13081e15ba8a29 : FAILURE in 40m 15s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/9e1104e2c0b74d118d4997457216c472 : FAILURE in 57m 08s","accounts_in_message":[],"_revision_number":3},{"id":"077cc7bae46b27db65819a6acbf8fa27e51ed8d7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-12-06 11:47:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2324b0daf349e28a95afa00ab00251711b267ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 13:00:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58e72505ab6f4f5b8ce844ddad3a6913\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/407b024af82e40fca8b334d79e3d30e4 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8820f89cff324bb8a36a3407273abd32 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d86a5d92af1d4297a396407d2ac94f98 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7094b08dba81460d9c6b41e01a14477c : SUCCESS in 7m 54s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/38cb1738baa14ee5b3299ffd7e0be1ee : SUCCESS in 32m 52s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/bb6d5363f72d4155b7148e0622daa152 : SUCCESS in 30m 58s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/024dddc623b844859c631244e5b703df : FAILURE in 41m 54s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/29fe795ebf50489bab1cc2334c97395f : FAILURE in 32m 56s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/5157ca87c0cd44f19ec949e758725e58 : FAILURE in 42m 02s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/8098efe192114719b601378e24fc472b : FAILURE in 1h 06m 50s","accounts_in_message":[],"_revision_number":3},{"id":"58aa92a992be3a40de47574f946549255f84c500","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-12-21 12:30:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1f1f3a71b5f5959ce4da88f00fed8c7ed9c5f124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 13:27:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dd227d00d5a43d0951f16ba2eee207d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec2ceb059e6f4b6580cad53fa1db2a13 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85f6ee2775a146b4a45bd08ec04c5bcb : FAILURE in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ee880e4a6b4424484893cec13cc7695 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccb92f70fc54ef2875e8127519fe12d : SUCCESS in 7m 55s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3cdd566153964bc89ff7d7c41c1b2f18 : SUCCESS in 41m 07s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1a08cb6078b8460b944ae3c53696447d : SUCCESS in 43m 55s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/65c0c677abfd4cc4a5abe64534873744 : SUCCESS in 42m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a1e52bd8d19748ed87ea56b9a7c95bb8 : SUCCESS in 49m 38s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/b6bf9a433fa04620bfb87b049f5c71be : SUCCESS in 32m 29s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/fca5f69db03548f69853b1b79ce02754 : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":3},{"id":"52aacb0374df93aa948daf31f0389859f097500c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-12-21 15:51:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a17113ab1ccedf2703ca120c1da76b6a5e399e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 18:06:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2ce1473ce2342ab99d54731742d2065\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4523e20d44eb498fb7790a5b6089e8fb : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4429cb69ebca4385af1f138fbe11fab8 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ce4b8f39699459bae56a23ca12a01ef : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b450889a034b4e88b5093d882ecd36 : SUCCESS in 7m 28s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b83f18a4e83b450680ea1a9795d47ae1 : TIMED_OUT in 2h 13m 57s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a9a5fd221bd6435b80cdf5baf2434e8b : SUCCESS in 1h 19m 34s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/080e15146dc64fb6b70a9eff0135c39d : SUCCESS in 53m 15s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/504ee1f34a6f47929d6f8775c510acfe : SUCCESS in 37m 49s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/adba84234b534ba588b89a4bb9e6a7b8 : SUCCESS in 1h 00m 37s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/443f7ae936f645acb025ec964248bd40 : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a2f9b468827c5f56e9a307847cb4556446d5c1c3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-12-22 10:05:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d89090ca6ead3dbcd39eb4015bf373e23c5ac9e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 11:12:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3d34f6e2c07431387a72f7cda3f360c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6dc96959f754c39a8a80733c2917589 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62cef76ad0984e898e95dc05b54507fa : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b7e5f58d9a94e41bffb42b8d9ce480b : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/881e47c3483740f59f52c3ddd4d8eb5d : SUCCESS in 8m 38s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c17e82afbc6f4ee19e56f7bab23c1901 : SUCCESS in 53m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5b58e417861c4a84a51a40f053289463 : SUCCESS in 41m 22s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e6fb0da96c304c49bd5874a0023a7a67 : SUCCESS in 43m 42s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4cf1611134c3476eb1d72e6565c86c23 : SUCCESS in 56m 24s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/fcddabdf8b494f95b58de29830b3d893 : SUCCESS in 1h 00m 50s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/f6a00a9946a3464ebc0feba77b700df1 : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":3},{"id":"a840ca3abcd1f083c856bc3b3a9ab861eed99346","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-25 12:17:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ef67dc3982547594524f19fb79da167f9e5fa838","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2024-02-12 11:53:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9930fc3ba767d997583c1389b58d524209c87a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-02-21 11:14:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9dcd2ca7a70d39c68694d1d77c0be668c2c19445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 12:23:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b258f6e8b7a848099c7667fda8261241\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af7ca06c9eb45d998fe98fca49f42a9 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83c339e1785d4491a2cbf58c09c3a331 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89943680e778496385c1851dbc26cf86 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5c428c5d494a9eb527fdeb24a90c8d : SUCCESS in 7m 27s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2897cff25a3a44aa97571290c93a17fd : FAILURE in 43m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e914f809eb6d4fe5be0a62a1ce559588 : FAILURE in 46m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4a5cba4ab55742d0bfb3daaab8a8f2e8 : FAILURE in 46m 28s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f1a5d69046944148af196b91bcd5f74d : FAILURE in 38m 16s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/c37330437d884f18bd155341a4a3f91f : FAILURE in 31m 56s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/7d311bef2fac4abbb4f08b384bd1f974 : FAILURE in 58m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ef9361c9416ccd1007da913489692e7d7139d9d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-02-21 17:47:07.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a274e8b00575edcd6bad949f1be26b423ba406d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 19:15:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb245fc6180c4268bb9041ffa14a6048\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57ee015c8ee4088b5bd119b0d7939bb : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cad86117c1b4d6cafd5553c132fc739 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0fe8b268c9d4ba8a75daf989e516c24 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2811976d540f4b4f9cbe3fd16289ebab : SUCCESS in 6m 20s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6b40cef40d3b404e8fddf4ab11d39975 : SUCCESS in 47m 31s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1a38c4039fae44069bdb0b6c574e1b00 : SUCCESS in 41m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/782bd4d3771142fb88cf2978a1c422db : SUCCESS in 59m 36s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/479815be453e4696a92bf3fb8c7c1574 : SUCCESS in 50m 51s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/810e296d6f814e979f8bad64b5b8df23 : FAILURE in 1h 01m 28s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/afc84cd5b3f0428784269c35bb48bf23 : SUCCESS in 1h 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ff5282655d505e54f98db83e4a9cb08bd0a8307a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-02-28 10:18:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"98eb49480557873d9d5d3ad810c24214b5935f29","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2024-03-07 15:50:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"074909cb2d18678ffcb52184dd026e78e6f625fa","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-03-12 16:23:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fb4a08f562385095013b1d416ca22f0beedecd3f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-04-12 08:50:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ad537486f91e675aa6a808fc5c05266d27bd71c3","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2024-04-17 16:16:07.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"973f3c2a9c0ae34f90f5590f5b009b7a1a028fb5","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-05-29 12:27:57.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0e5a0e41879bdb67862c8c43a61a878a56663cbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 12:28:19.000000000","message":"Patch Set 5: Verified-2\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/aodh on branch master.  The\nproblem was:\n\n  Job telemetry-tempest-base not defined\n\nThe problem appears in the the \"aodh-tempest-plugin-postgresql\" job stanza:\n\n  job:\n      name: aodh-tempest-plugin-postgresql\n      parent: telemetry-tempest-base\n      irrelevant-files: *aodh-irrelevant-files\n      vars:\n        devstack_services:\n          postgresql: true\n          mysql: false\n        devstack_localrc:\n  ...\n\n  in \"openstack/aodh/.zuul.yaml@master\", line 105","accounts_in_message":[],"_revision_number":5},{"id":"7e7f37126a6bfabdb8cce5da20284c79f8f572cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-05-29 12:40:28.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4ba45030702f16d1da95f2b6ea91aaf112a99f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 12:40:50.000000000","message":"Patch Set 6: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/aodh on branch master.  The\nproblem was:\n\n  Job telemetry-tempest-base not defined\n\nThe problem appears in the the \"aodh-tempest-plugin-postgresql\" job stanza:\n\n  job:\n      name: aodh-tempest-plugin-postgresql\n      parent: telemetry-tempest-base\n      irrelevant-files: *aodh-irrelevant-files\n      vars:\n        devstack_services:\n          postgresql: true\n          mysql: false\n        devstack_localrc:\n  ...\n\n  in \"openstack/aodh/.zuul.yaml@master\", line 105","accounts_in_message":[],"_revision_number":6},{"id":"6606c359893315e5dcf68fca367940d947c2ed2f","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-06-04 10:59:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"019f8a0958713d5c78691fae5b362f52bc37aea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 12:26:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c0d346cb7a640fdb44fa581173c9ee2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e384b90607d04c8ebc9c59f765b61767 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/260071810a8348a3b6bacf2119593d63 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca54d399ddb84c1fa95a10da1a800301 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a6b3e0b9cb64e19a18060db4651e2c5 : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2662450bab4097aedf51cad223d9a9 : SUCCESS in 6m 40s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b98f14c823754f2d8fc048f61e513b1b : SUCCESS in 41m 42s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/fd2b99dfb65c4b7fadc29a09c49d8f3b : SUCCESS in 31m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/02ef5ae851f643cbaadbb49bd3b20a2f : SUCCESS in 47m 34s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2026991314414996923a7e8829687049 : SUCCESS in 40m 59s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/93e6cea666b84d7f8f80605f0f01f496 : SUCCESS in 59m 03s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/4d714daf0d0644bcbed5bf893ff6c1c2 : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":6},{"id":"55f6635019bf5c82f2c991041b38167de505199d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-06-04 12:52:55.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"64b177be8a246232fc819da71c49a93a99c58b01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 12:53:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c7c06cba20f0bf45510db533dc3bf7eb8d00c426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 14:05:26.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b3f7e647965402f9094102b63aab5ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366d1d850436463cbdf5e2a6f94a6723 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fff12f30f67d4160930f7ab67031147f : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c201ac79339461abe3bf528a1fcb040 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b422085efe4267af2e9e67f25caa31 : SUCCESS in 6m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2af7750316f646cd80194518ebcd2b44 : SUCCESS in 39m 43s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e0212459b4a649ea8630282fde9f835b : SUCCESS in 43m 55s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/53b34fca6f90420d86107cadad8d272f : SUCCESS in 53m 20s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d863868c1e204b95895b6c0f88798d0f : SUCCESS in 53m 29s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/955701a603ae4596b7d6afc01a65f0f2 : SUCCESS in 59m 48s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/5695cb13cfce453c95f18edb9485360e : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":6},{"id":"3319573494a83f6b12cbffcf6ca3d2be392ce663","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 14:05:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"69894a3bc2d6e06230f7347bbfed31837f21221a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 14:07:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3ffb4d6163f488fae90f6fc4ed53731\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2cf173a6f124ee5beeb406774479294 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bd066696276c7cd60f443b3bcb3e1b637c5a8a74","revisions":{"411285d6eee3db2a5b3ce879cbe22647d104dbd7":{"kind":"REWORK","_number":1,"created":"2023-12-01 11:49:11.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/74/902374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/74/902374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/74/902374/1"}}},"commit":{"parents":[{"commit":"59a43ef35a9a0058b182ee49180857ada9dba9a3","subject":"Move functional tests to tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/59a43ef35a9a0058b182ee49180857ada9dba9a3"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-01 11:48:11.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-01 11:48:11.000000000","tz":0},"subject":"Add tempest configs to jobs","message":"Add tempest configs to jobs\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/902370\nChange-Id: I5f1fe1a02dbccfca48c95e97657c72054016a17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/411285d6eee3db2a5b3ce879cbe22647d104dbd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/411285d6eee3db2a5b3ce879cbe22647d104dbd7"}]},"branch":"refs/heads/master"},"b7fea762ae855d9c2b0cd5e8cc603c9c5c84e1c0":{"kind":"REWORK","_number":2,"created":"2023-12-06 10:08:42.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/74/902374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/74/902374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/74/902374/2"}}},"commit":{"parents":[{"commit":"59a43ef35a9a0058b182ee49180857ada9dba9a3","subject":"Move functional tests to tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/59a43ef35a9a0058b182ee49180857ada9dba9a3"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-01 11:48:11.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-06 09:54:39.000000000","tz":0},"subject":"Add tempest configs to jobs","message":"Add tempest configs to jobs\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/902370\nChange-Id: I5f1fe1a02dbccfca48c95e97657c72054016a17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b7fea762ae855d9c2b0cd5e8cc603c9c5c84e1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b7fea762ae855d9c2b0cd5e8cc603c9c5c84e1c0"}]},"branch":"refs/heads/master"},"7bff9c745b1dbf12b67c4e6e42e35f79e8fbe303":{"kind":"REWORK","_number":3,"created":"2023-12-06 10:20:03.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/74/902374/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/74/902374/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/74/902374/3"}}},"commit":{"parents":[{"commit":"59a43ef35a9a0058b182ee49180857ada9dba9a3","subject":"Move functional tests to tempest_plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/59a43ef35a9a0058b182ee49180857ada9dba9a3"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-01 11:48:11.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-06 10:19:50.000000000","tz":0},"subject":"Add tempest configs to jobs","message":"Add tempest configs to jobs\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/902370\nChange-Id: I5f1fe1a02dbccfca48c95e97657c72054016a17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7bff9c745b1dbf12b67c4e6e42e35f79e8fbe303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7bff9c745b1dbf12b67c4e6e42e35f79e8fbe303"}]},"branch":"refs/heads/master"},"842aef3e55f7f3715f01016fff38120cb68784cd":{"kind":"REWORK","_number":4,"created":"2024-02-21 11:14:34.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/74/902374/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/74/902374/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/74/902374/4"}}},"commit":{"parents":[{"commit":"64e0f7c895100ae9ee6379f5bd27f1121c26d96c","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/64e0f7c895100ae9ee6379f5bd27f1121c26d96c"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-01 11:48:11.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2024-02-21 11:13:28.000000000","tz":0},"subject":"Add tempest configs to jobs","message":"Add tempest configs to jobs\n\nChange-Id: I5f1fe1a02dbccfca48c95e97657c72054016a17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/842aef3e55f7f3715f01016fff38120cb68784cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/842aef3e55f7f3715f01016fff38120cb68784cd"}]},"branch":"refs/heads/master"},"ea3bcae4fef597d4c50c28cba188801a8110c1cf":{"kind":"REWORK","_number":5,"created":"2024-02-21 17:47:07.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/74/902374/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/74/902374/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/74/902374/5"}}},"commit":{"parents":[{"commit":"64e0f7c895100ae9ee6379f5bd27f1121c26d96c","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/64e0f7c895100ae9ee6379f5bd27f1121c26d96c"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-01 11:48:11.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2024-02-21 17:46:57.000000000","tz":0},"subject":"Add tempest configs to jobs","message":"Add tempest configs to jobs\n\nChange-Id: I5f1fe1a02dbccfca48c95e97657c72054016a17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ea3bcae4fef597d4c50c28cba188801a8110c1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ea3bcae4fef597d4c50c28cba188801a8110c1cf"}]},"branch":"refs/heads/master"},"bd066696276c7cd60f443b3bcb3e1b637c5a8a74":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-29 12:40:28.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/74/902374/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/74/902374/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/74/902374/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/74/902374/6"}}},"commit":{"parents":[{"commit":"5e2851df640c04dffaef6fb043d63be357beae95","subject":"Merge \"Remove SQLAlchemy tips jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5e2851df640c04dffaef6fb043d63be357beae95"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2023-12-01 11:48:11.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2024-05-29 12:40:28.000000000","tz":0},"subject":"Add tempest configs to jobs","message":"Add tempest configs to jobs\n\nChange-Id: I5f1fe1a02dbccfca48c95e97657c72054016a17e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bd066696276c7cd60f443b3bcb3e1b637c5a8a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bd066696276c7cd60f443b3bcb3e1b637c5a8a74"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
