)]}'
{"id":"openstack%2Faodh~905066","triplet_id":"openstack%2Faodh~stable%2Fyoga~I88b803fcbd7a4b7a26110f7ef55aa5885e2792a4","project":"openstack/aodh","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-01-25 12:32:35.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-01-16 09:39:13.000000000","reason":"removed on reply"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2024-01-25 12:32:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88b803fcbd7a4b7a26110f7ef55aa5885e2792a4","subject":"Pin gnocchi to 4.4.5 in Yoga","status":"MERGED","created":"2024-01-09 07:55:32.000000000","updated":"2024-01-25 12:33:34.000000000","submitted":"2024-01-25 12:32:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"905066","meta_rev_id":"885affd39ecdb7eb22e3b3a71f5fa04cb44cb2ab","_number":905066,"virtual_id_number":905066,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 12:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-01-16 09:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2024-01-25 12:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-09 08:27:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-09 08:29:03.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-01-09 08:29:03.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"beaa971ea116e6ef01e41210099aa71bcbd2f6fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-01-09 07:55:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53f4e7a88f93b69dcba2731a90633b507798ea33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 08:27:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a86894df9064f589433b91ff36aa6c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee9c54ae3df415c836ab0cc8c131a65 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1eb774621bc4e1d98b05d43d79622aa : SUCCESS in 8m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c059cc4db94363b585bf736b817734 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10265182a3e2476699dc4e3371bda01e : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2a044564dfcbb0e536d6cf1f1132a5221f0cd785","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-15 07:25:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8a19fb9e584a0b08bfa0acc37530038f9c02bac","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-01-15 12:33:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d73aabd136e56b724f27ee106b701b2c416fcd72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-01-15 13:13:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b178f034e0c0ee5bb39396f5ae60dd52867bf153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:44:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/476370d1663e452f8e11f760ce0b0912\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9637f63a1c847dbb16a12a03c10958b : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf599eb96fd41cbbc9f6b52264a32ad : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc80e1db25dd42f1aa6e3fbfb911cf3b : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0537b1cbb71e4d33b23b2689d768c411 : FAILURE in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae128626467a4d5bbc0ee98a07478455 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7ef09b267dee975b1c9945027d6e7a6c2dfdd428","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-01-15 13:48:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a36b2c5d9cc067b2a834fbfde8b01391efc576b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 14:16:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfe8f51ca98047ae9b4b103f72f7f9c9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb01c19fba774196995f147990e81978 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2f7c36391174bff85fc66f04f8fce88 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8bab5f30aa5448fa11ddbfb7c34bf80 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9bc5ec9ea134a0b951d6ad45b629acb : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c147aad42ec04cd88781e20312fa9d0d : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d6ba529455e76e7edc22560cc02c0b9e1f0f45bc","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-01-16 08:59:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c31e21b18c7a6587e240c7f26e79b3a9655d135","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-16 09:39:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"729b8b0320302b14aa000a6094cee16fb5f9bcce","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-25 12:14:30.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2479e6ac73292d9b4431ccb6e1c06901bcd14796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:14:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a8f40273fb42b717c973231d84afd1050928a4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:32:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34269610f0d5474aa10c2d7c86f1bccc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51eef547837a4f799c75a6a03dd6a093 : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b8f06398e204b0ea2f0c9b96bf5592c : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d418f53bf0904422aaa4d57b0c9c27a6 : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b43b12d42c6849fbaa4cf69100e9bec2 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab70ec9f93a4cba93bc67a7c67be287 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":2},{"id":"2ef7a85aa5803abbf4f35303c70c71adc786cab1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:32:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"885affd39ecdb7eb22e3b3a71f5fa04cb44cb2ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:33:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8edf17af8bae44ba9a1356b9dc6f2d31\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b2e8363a51437288b55d5291c6a022 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"25f72e06535cdcf95f1300652f8c05e62a77b989","revisions":{"faf8a3eeecec20e7752131c2a34a73089c829a51":{"kind":"REWORK","_number":1,"created":"2024-01-09 07:55:32.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/66/905066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/66/905066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/905066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/905066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/905066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/66/905066/1"}}},"commit":{"parents":[{"commit":"f9abe01eb14416474c2bc3c15311b5235adbf661","subject":"Merge \"[Yoga] Remove TripleO job\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f9abe01eb14416474c2bc3c15311b5235adbf661"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-01-09 07:50:56.000000000","tz":-300},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-01-09 07:50:56.000000000","tz":-300},"subject":"Pin gnocchi to 4.4.5 in Yoga","message":"Pin gnocchi to 4.4.5 in Yoga\n\nThis fixes the CI, which is currently failing. It pins gnocchi\nto 4.4.5, which is the same version as is used in ceilometer\nCI on yoga\n\nChange-Id: I88b803fcbd7a4b7a26110f7ef55aa5885e2792a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/faf8a3eeecec20e7752131c2a34a73089c829a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/faf8a3eeecec20e7752131c2a34a73089c829a51"}]},"branch":"refs/heads/stable/yoga"},"25f72e06535cdcf95f1300652f8c05e62a77b989":{"kind":"REWORK","_number":2,"created":"2024-01-15 13:13:32.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/66/905066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/66/905066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/905066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/905066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/905066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/66/905066/2"}}},"commit":{"parents":[{"commit":"f9abe01eb14416474c2bc3c15311b5235adbf661","subject":"Merge \"[Yoga] Remove TripleO job\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f9abe01eb14416474c2bc3c15311b5235adbf661"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-01-09 07:50:56.000000000","tz":-300},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-01-15 13:09:30.000000000","tz":-300},"subject":"Pin gnocchi to 4.4.5 in Yoga","message":"Pin gnocchi to 4.4.5 in Yoga\n\nThis fixes the CI, which is currently failing. It pins gnocchi\nto 4.4.5 for testing, which is the same version as is used\nin ceilometer CI on yoga\n\nChange-Id: I88b803fcbd7a4b7a26110f7ef55aa5885e2792a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/25f72e06535cdcf95f1300652f8c05e62a77b989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/25f72e06535cdcf95f1300652f8c05e62a77b989"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
