)]}'
{"id":"openstack%2Faodh~906945","triplet_id":"openstack%2Faodh~master~If6aeb6c837d413c9d9a544a80c4c736efc8455fc","project":"openstack/aodh","branch":"master","topic":"remove-pytz","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2024-06-05 14:10:51.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-06-05 10:41:42.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-06-05 14:10:51.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-06-05 14:10:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If6aeb6c837d413c9d9a544a80c4c736efc8455fc","subject":"Use zoneinfo instead of pytz if available","status":"MERGED","created":"2024-01-27 10:38:01.000000000","updated":"2024-06-05 14:11:55.000000000","submitted":"2024-06-05 14:10:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906945-remove-pytz","meta_rev_id":"a892d268405105f335074c90656829c72965a1b7","_number":906945,"virtual_id_number":906945,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-05 14:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2024-06-05 10:41:42.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-01-30 08:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2024-06-05 10:41:42.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-27 12:27:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-27 15:02:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-30 08:30:22.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-03-15 03:41:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-03-15 03:41:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"}],"messages":[{"id":"0806c27f75ed798d9a373f0497f68414ca7ec928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-27 10:38:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11f2ebe905b48535934164ce3f4ca4200a50ea2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-27 11:26:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"effdbee8a48a8f32290a5dc74dbae72949dbee8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 12:27:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d16b5128d3e4437b98b205911560254\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3593f9f30cf74e31be0453026e5b50d2 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bed97dd144d443c9d08bca50baf8c97 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/639c4e6295c548dbbd759364e5f435cb : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e04c86d0544053a457535f104bf5f7 : SUCCESS in 7m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a90f3a9e42bd4ff981cda62acae25c4d : SUCCESS in 2m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4d8afcc2a663453fbdd72991bdffe7b6 : SUCCESS in 47m 57s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e3452bc8b6144d22b6dedf748efe9e4e : SUCCESS in 48m 44s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/904461ca433e4ba9a4f25a8c65b563e5 : SUCCESS in 31m 48s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/48f9410d80f7418ba7d0140cfe4aa634 : SUCCESS in 52m 41s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6c13c7bc0b5c434cb911950350498796 : SUCCESS in 55m 32s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/e0dec6dbad8746529d024ff1ef49619f : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":2},{"id":"42610e721e5500a732afb3df88f61c7ed3b9850c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-30 08:30:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1342ad35ada6f3503281ee18e921f0e61d7de6d3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-04-29 10:34:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ecc6a3b19893906052ce490ef75322da5afe7ad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-30 04:11:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7cbad9155f3b5fe1f485ac36fce841e424ea1c0b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-06-05 10:41:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6bb966b2de09c59f7c770a8e5cf15f7f2d463716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 10:42:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a08a35f1f27de4f44eb6cb5a5145594d3b8ca9b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 11:51:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80e1a0b6e1ed4269b81eb6da38110d66\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe11d4775234b13b861446db004569e : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2052a9bb2a284f5eb1ac9ada94a697a9 : SUCCESS in 7m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e70cfd17701491d94f59e0e961c128b : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836d197332514ba9b41601092123d20c : SUCCESS in 8m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c331a3c03c2a4963bfc3de368f5d6368 : SUCCESS in 2m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/49555beae02747e696dc5c7bf1f15c05 : SUCCESS in 45m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/55e5c88c5b9747a0885691c159d43913 : SUCCESS in 35m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8608f382d7304fcfa2e1b4129b104fdb : SUCCESS in 51m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3f872b842454427a8e374e798ca6a244 : SUCCESS in 56m 44s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/8f5ddebebfd947a8ac1d6779421a9e6a : SUCCESS in 1h 03m 15s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/b6f1af4fa5b14448a7e2cd123d97a97a : FAILURE in 57m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c62096007cc6c7157b17605601957d95d72d58c6","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-06-05 11:57:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3abc73776cf33fdb099cb0fe2fd67c746b3e042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:57:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ed72b70a0914ed4848086ce5c8124a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/583114396a804980816d92c7c9db87a5 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13da9aaad3244a8b912f25961d3a3767 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5f4cc28cb4643bfafa74ec8baea3051 : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b46183bc8cf74c0f8df2ca838101edbd : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cae2c3af10e42088d472110f09e1b9e : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/03c605a7fc604f08893038e73d9e342e : SUCCESS in 2m 23s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5c22478449a6437b903dae32abd11aca : SUCCESS in 45m 41s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/76acd520f2084076a1526d7d395256f4 : SUCCESS in 33m 21s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c961bc7d658a4051940e2ae9e40dbfe9 : SUCCESS in 49m 57s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/56c362109c8d4ac0a85fee19cd15fc55 : SUCCESS in 45m 03s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a5dea2e4a9d5468f929599e6779e083a : SUCCESS in 53m 46s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/3c5ef009f104443da9f9ae998e2d7ff4 : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":2},{"id":"36937c32b22ccd7fcde0c34f3abca668f5798051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:58:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"44218c1ea293fc6f5aeeee064c9f373f27cd1496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:10:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00c196836a0242fbb9e81c0151dbacdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a8d0fbd7dc42b4b84b657d57eade2c : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff607c17ad234b7badb27bb80505db9c : SUCCESS in 8m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5d0e0e123f84f4d9c2c8713642935de : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60783f9a34ab4aa687a25b57796c9b7a : SUCCESS in 9m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/76f688bd3ee243ddbd73508a9525cc0e : SUCCESS in 3m 32s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e96a442986f84cee81172732aa8815e3 : SUCCESS in 33m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/827205f894a64bcbbd51f250154f5483 : SUCCESS in 30m 35s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ece99728525c44408fbc86fd6ac671a1 : SUCCESS in 40m 14s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1a9a27ab5a224f1ba4d53e6ad92536b3 : SUCCESS in 38m 21s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/07697bbe86bf4fbab4b2565584887b06 : SUCCESS in 1h 07m 32s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/328c4de8047e41968bb3d254d5119857 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9314ef4ae0ac31d4ed81520cd5bac45d73f85f8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:10:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a892d268405105f335074c90656829c72965a1b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:11:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e1c7b6ae054cc884f8bef84d8ba01e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dcf67ddefe84ba087a768182fa5ccfd : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b312d0a5ce99aa6dd434e3384eaefb1b807b6bd","revisions":{"2586c4f26d472b34b406b911b2c8346b4d8f063f":{"kind":"REWORK","_number":1,"created":"2024-01-27 10:38:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/906945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/45/906945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/45/906945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/45/906945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/45/906945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/45/906945/1"}}},"commit":{"parents":[{"commit":"3d49da3ce6a8043bf383ee7a90cfcc04cee4f06b","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3d49da3ce6a8043bf383ee7a90cfcc04cee4f06b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 10:37:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 10:37:01.000000000","tz":540},"subject":"Use zoneinfo instead of pytz if available","message":"Use zoneinfo instead of pytz if available\n\nThe zoneinfo module was introduced in Python 3.9. We can use that\nbuilt-in module instead of the 3rd party library (pytz).\n\nThis is based on change I1f88bdadc68bfa726eac1da1c5824c1ed352ad98 in\noslo.utils.\n\nChange-Id: If6aeb6c837d413c9d9a544a80c4c736efc8455fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2586c4f26d472b34b406b911b2c8346b4d8f063f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2586c4f26d472b34b406b911b2c8346b4d8f063f"}]},"branch":"refs/heads/master"},"6b312d0a5ce99aa6dd434e3384eaefb1b807b6bd":{"kind":"REWORK","_number":2,"created":"2024-01-27 11:26:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/906945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/45/906945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/45/906945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/45/906945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/45/906945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/45/906945/2"}}},"commit":{"parents":[{"commit":"d2c2e1b671a6f4d58bd43162018ddd0df011a5de","subject":"Catch specific exception when validating time zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d2c2e1b671a6f4d58bd43162018ddd0df011a5de"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 10:37:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 11:26:39.000000000","tz":540},"subject":"Use zoneinfo instead of pytz if available","message":"Use zoneinfo instead of pytz if available\n\nThe zoneinfo module was introduced in Python 3.9. We can use that\nbuilt-in module instead of the 3rd party library (pytz).\n\nThis is based on change I1f88bdadc68bfa726eac1da1c5824c1ed352ad98 in\noslo.utils.\n\nChange-Id: If6aeb6c837d413c9d9a544a80c4c736efc8455fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6b312d0a5ce99aa6dd434e3384eaefb1b807b6bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6b312d0a5ce99aa6dd434e3384eaefb1b807b6bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
