)]}'
{"id":"openstack%2Faodh~907184","triplet_id":"openstack%2Faodh~master~I963554a35a6aa2e893ba199975201bc2a1e15429","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2024-02-13 15:29:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I963554a35a6aa2e893ba199975201bc2a1e15429","subject":"Use SHA256 hashing algorithm instead of MD5","status":"ABANDONED","created":"2024-01-30 09:24:27.000000000","updated":"2024-02-13 15:29:13.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"dfb1149dcaf0369fbf88fbb33338d59ecf5b280d","_number":907184,"virtual_id_number":907184,"owner":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-30 10:35:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":-1,"date":"2024-01-30 10:44:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":-1,"date":"2024-02-01 08:11:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":-1,"date":"2024-01-31 08:16:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-30 09:52:27.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-01-30 10:35:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-30 10:44:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-01-31 08:16:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"5e3e737299cc51a8eae6d94dfa182301a95ac14e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2024-01-30 09:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bf9244943a3f23eb01f7bdef4c245056d1d12fb","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-30 09:52:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd50a6b322d812d7da9438a706c0f7860044e2b2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-01-30 09:52:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d295b8fa4244da52e267ac481be6268de9922572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 10:35:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fb228093ce8404bbf8f26d0af8bc1a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f0d0ba4137424c8a1a46d8b7193b86 : FAILURE in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1eceeebc9bdd46158cf3185b3ed6727c : FAILURE in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e1a63889d854f78b3c4c1396021f2d6 : FAILURE in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4ee469f15e40d7a92cc4c402a128f2 : SUCCESS in 6m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5040bc4542a1457ca1735f8323b66c10 : SUCCESS in 27m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b41e3e2d344c4e2aac72fc405f800faa : SUCCESS in 34m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9e56014a38ad4558b4f52b0fcb75e20c : SUCCESS in 51m 22s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/849bd9d4bb4f49b48010936a5f4cabec : SUCCESS in 50m 15s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/8d8614c8a2f0400a928743ffd0f3cbc4 : SUCCESS in 42m 05s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/28c9288705004ac39561f462ab3d5bf4 : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":1},{"id":"111675a71572bc481e7b7a149f1cd95819447b4a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-30 10:44:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f48644d5f169081c9fa178f6ba5d7da88f5cd8c1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-31 08:00:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df4246216fb8937f8f3c547ae8368ebfb1bae581","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-31 08:16:09.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"3eab6bb6be9226aef67f5a60bfd5c5db474c5029","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-02-01 08:11:09.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"dfb1149dcaf0369fbf88fbb33338d59ecf5b280d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2024-02-13 15:29:13.000000000","message":"Abandoned\n\nAbandoned in favour of https://review.opendev.org/c/openstack/aodh/+/907296","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41d43333b9bd48a804a3f3a461696756702cd72f","revisions":{"41d43333b9bd48a804a3f3a461696756702cd72f":{"kind":"REWORK","_number":1,"created":"2024-01-30 09:24:27.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/84/907184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/84/907184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/84/907184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/84/907184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/84/907184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/84/907184/1"}}},"commit":{"parents":[{"commit":"3d49da3ce6a8043bf383ee7a90cfcc04cee4f06b","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3d49da3ce6a8043bf383ee7a90cfcc04cee4f06b"}]}],"author":{"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","date":"2024-01-30 09:24:00.000000000","tz":330},"committer":{"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","date":"2024-01-30 09:24:24.000000000","tz":330},"subject":"Use SHA256 hashing algorithm instead of MD5","message":"Use SHA256 hashing algorithm instead of MD5\n\nSince md5 is not fips compliant, aodh reports error\n\"ValueError: [digital envelope routines] unsupported\" after creating\nalarms.\n\nUse sha256 instead which is compatible with fips enabled system and\nalso has several advantages over MD5 and SHA-1.\n\nChange-Id: I963554a35a6aa2e893ba199975201bc2a1e15429\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/41d43333b9bd48a804a3f3a461696756702cd72f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/41d43333b9bd48a804a3f3a461696756702cd72f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
