)]}'
{"id":"openstack%2Faodh~907617","triplet_id":"openstack%2Faodh~master~I980ef8bca297bace734c73c4bcc3a43e7da4c0e2","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-05 11:24:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I980ef8bca297bace734c73c4bcc3a43e7da4c0e2","subject":"Skip EOM branches for release note builds","status":"ABANDONED","created":"2024-02-02 16:16:10.000000000","updated":"2024-02-05 11:24:07.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"489e1132129c926a680edf3286e16f7d4cbbdf4a","_number":907617,"virtual_id_number":907617,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-02 17:32:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":-1,"date":"2024-02-05 11:11:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":-1,"date":"2024-02-02 16:18:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-02 17:32:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 11:11:14.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"4bb9c6fa968ac8d1a8335c523b2f3fc997353ff0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 16:16:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00b12133f50898c5cfe12ef700be81d881c81d52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 16:18:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"06032c83344aa45addc5e21847a762f5d5fd2c2e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 16:18:55.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"03acba37c87c8991b4c15f0a3d4c902c37a04bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 17:32:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220dc4cda4ec425d9ff67de47aea3603\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa3f9d01c5b49b984f0995aa4628937 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4d46ba6a4ea4f3d980510acea4486dd : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84e600395b5b43f886622fc710ae94c2 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f3eb9da3dd4a77a3bd5c149cb552b3 : SUCCESS in 7m 21s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/197cac77ef754715b3629bc16264d52a : SUCCESS in 41m 16s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1782963923f14d99817702ee143f04b1 : SUCCESS in 48m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f895ceee81c544cfa7562bfa284552aa : SUCCESS in 43m 30s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/30b2b801dc24467f94b949bec53f4e4c : SUCCESS in 49m 24s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/1f6fd6b4cae14404b54c8fbf9ca6eb62 : SUCCESS in 1h 04m 39s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/fb915ffd3ccf429ca832fdd3e01e34ff : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":2},{"id":"fe55e198420879c9c0f999df7485f93e1e348e08","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-02-05 11:11:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"489e1132129c926a680edf3286e16f7d4cbbdf4a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 11:24:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0f5ad117c70d8239f85fd0227d39cb82c59006e","revisions":{"b0577a52e8b5e4e9ff043c52769f7dcfc22206b8":{"kind":"REWORK","_number":1,"created":"2024-02-02 16:16:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/907617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/17/907617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/907617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/907617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/907617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/17/907617/1"}}},"commit":{"parents":[{"commit":"3d49da3ce6a8043bf383ee7a90cfcc04cee4f06b","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3d49da3ce6a8043bf383ee7a90cfcc04cee4f06b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 16:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 16:15:52.000000000","tz":540},"subject":"Skip EOM branches for release note builds","message":"Skip EOM branches for release note builds\n\nChange-Id: I980ef8bca297bace734c73c4bcc3a43e7da4c0e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b0577a52e8b5e4e9ff043c52769f7dcfc22206b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b0577a52e8b5e4e9ff043c52769f7dcfc22206b8"}]},"branch":"refs/heads/master"},"d0f5ad117c70d8239f85fd0227d39cb82c59006e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-02 16:18:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/907617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/17/907617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/907617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/907617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/907617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/17/907617/2"}}},"commit":{"parents":[{"commit":"3d49da3ce6a8043bf383ee7a90cfcc04cee4f06b","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3d49da3ce6a8043bf383ee7a90cfcc04cee4f06b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 16:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 16:18:45.000000000","tz":0},"subject":"Skip EOM branches for release note builds","message":"Skip EOM branches for release note builds\n\n... to avoid the failure caused by stable/yoga, which was transitioned\nto unmaintained status.\n\nChange-Id: I980ef8bca297bace734c73c4bcc3a43e7da4c0e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d0f5ad117c70d8239f85fd0227d39cb82c59006e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d0f5ad117c70d8239f85fd0227d39cb82c59006e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
