)]}'
{"id":"openstack%2Faodh~916382","triplet_id":"openstack%2Faodh~master~Ief7ea81e78344ea32fb2a295f11a5d5a873251a8","project":"openstack/aodh","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-05-08 12:35:26.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-09 12:18:38.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2024-05-09 12:18:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief7ea81e78344ea32fb2a295f11a5d5a873251a8","subject":"Fix db upgrade with SQLAlchemy 2.0","status":"MERGED","created":"2024-04-19 07:04:32.000000000","updated":"2024-05-09 12:19:43.000000000","submitted":"2024-05-09 12:18:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916382-sqlalchemy-20","meta_rev_id":"eacb6098224ad1a66a1d269fe0a27a399db21645","_number":916382,"virtual_id_number":916382,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 12:18:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2024-05-08 12:35:26.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2024-05-07 14:45:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-05-07 14:31:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2024-05-07 14:50:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2024-05-08 12:35:26.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 08:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-22 09:42:07.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"CC"},{"updated":"2024-05-07 13:01:11.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"CC"},{"updated":"2024-05-07 14:31:27.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2024-05-07 14:43:44.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2024-05-07 14:50:08.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-05-08 12:35:26.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"6f55cdc97dbf99d0949e19db9755cbe4bf338436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 07:04:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac011f386dc6ee08b1a22640b3f99d2c76779007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 07:06:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a452e493c39551204f6bbd1a6eb02125c0694c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 07:12:36.000000000","message":"Topic set to sqlalchemy-20","accounts_in_message":[],"_revision_number":2},{"id":"31da8fed5ed243d404c57866dcafb07de0b84459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 08:17:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f13e15bb75c46d990f1d8369f5608d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/111af5c8f1db41fbbfdda4ee44dd874b : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4949cd58886646d89ceb6c4934aaad49 : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5a29ad4df12461089cf9cf83a89edd8 : SUCCESS in 7m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2dcfe59c815b4344a1af6457136ed4b4 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/627a302482b848868d4985ed9cde3163 : SUCCESS in 7m 14s\n- aodh-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f65229efd1e4c92965445b5d0d9fba0 : SUCCESS in 6m 15s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/31521fb7dbb046c0a765ac68bbe8fc28 : SUCCESS in 52m 10s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e48b01d8b10e4fc38c7827b8e050770a : SUCCESS in 43m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/229b35d5a51248c2b070daf1c8a161f5 : SUCCESS in 52m 53s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/70d8a2751b2643128509ccf21a4b3e4a : SUCCESS in 38m 14s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/3c12d9eb3e664072abbc2f2e021e80cb : SUCCESS in 44m 16s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/00bfa9b1a33142c48256e56479689480 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a3f15c5fac271f9d733da32b90ffb5f9b31e3b1a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 08:38:21.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"4ad664277f2d301050c5df4032b84b0df4643f11","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 08:55:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d24682dcf9a204b52ebd50452dd7dbfa3582e334","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-04-22 09:42:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97bfee042d030927255c511f97aa5e031fe317c3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-22 09:52:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fef609ab3a4a4661932c4f9d0e76a43ff1f30273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-22 14:20:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"78df8896eea1537abc0f6e6159e7bd4462b2b0f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:27:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/159ad75f75be4e9685bba671447b2fd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1aae1adc8ba460397d83755ac6b3934 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7818fdf8b197474abe49990bb5fb07a6 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1d33867e6754283921e1f5c708c2876 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d15cafe80e3a4eb9a9dee3fa850fbd0b : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f5d14bdce74d05ac76f473c2a7cb81 : SUCCESS in 7m 17s\n- aodh-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8a7d5d1327144afda02a2d084d38de96 : SUCCESS in 7m 07s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a4bc441948b140169da0a2400c660e18 : SUCCESS in 33m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0c733bce72344549a49dc010a77e40b2 : SUCCESS in 49m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4b0ba2cb3b5f45dd85211f00ea868b0b : SUCCESS in 53m 55s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d61969d518a14b90959c1b83c2b46282 : SUCCESS in 53m 39s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/3815bbea97c846509c0a5a7ca6d7cdf1 : SUCCESS in 54m 45s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/1201f972c2b0479c9474694fdfe2d9ea : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":3},{"id":"35bd59d4cb24ea9e5394b5fd16b8325b6427ca30","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-22 15:30:48.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"9c7cadfa594211ea04ed03776c9d6e4db56d8aee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-22 15:31:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83f0cbf5bdbd1819951586b67cf6392c3f515a48","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 13:01:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fccc8587ca1f47fa3516e99e2aed8deaacb9c5ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 13:04:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"96a2038a35f00cd72ab9f268737c9c76f01d0850","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 13:19:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a2cfc031ddaeae0889fe6e4ab6344b96ccf3d1f","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 13:24:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b76d8c5ed0bb135d6bf4773cbebba3bc0ec461a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 13:24:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f0b0bef38ee13e0707376eff6815f7642b18c8a1","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 13:24:53.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fb32f8991f54ab5c5c529269e1c320cc77b2561d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 14:29:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa7bd0e99b164480aa73ec5be4bbd4e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6150f7cd52cc4d389437c1dfdeeac46b : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/632e85292b974d25a088a379cdd93714 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/025f9be3b0a04cc4a2e87b9dc20cfa5b : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cb943453d034ed999eedb96f1cdddc9 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dad2582653a445bb7cd7aa63cc3530e : SUCCESS in 6m 46s\n- aodh-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b2a200d47f3a4b3cbfbe22f5d061f8f5 : SUCCESS in 6m 54s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/efe5b8cead1d4cc5b90299d1722a8f62 : SUCCESS in 36m 35s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/67884a691a66423f91ee00682e9809c5 : SUCCESS in 32m 08s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f30987e89b32494a92c07030d4a8627f : SUCCESS in 54m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/98902025b08c4ae7a3d33596a22830b1 : SUCCESS in 40m 11s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0f874b71fc5647f488c62b4957cbf056 : SUCCESS in 51m 44s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/4934aca808574639be724ca0349d972f : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":5},{"id":"c5a3208a2306cbc02ea8cff006bc0ec308906b6d","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 14:31:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"420777a5ead3e4d46bc65e1c04a1aedca4e4b5d9","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 14:36:24.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cc685898b797ced40fc1bffbf5bd62a7d6354f6d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 14:45:38.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3dcad07dc628d9750742d9e32a87ea6648c2dbc1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-05-07 14:50:08.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"552a9ef5b3d544ce7a0481c377710f2ad4c8906e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-05-08 12:35:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6f2374185466784eb61225c6ee9a55d4169a72c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 12:36:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f38d2e32ed2597cecedd8e5fdf363a4facdc3fad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 13:47:25.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9909451c631a48baa8ce74aabd3415b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47823fae478c43e2be919011f2fc421f : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6bcbe8eb3134dd68837f4fed6ef8ca0 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bf42046670c480baa49c1055bf62052 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6890cb09c7804b50aad9d15019887871 : SUCCESS in 7m 15s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a77fd6f981ab4eafb79cda7f1072b860 : SUCCESS in 46m 53s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/633a6bff4d084d6493e6d46c846e3c37 : SUCCESS in 48m 03s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2b01384424f44cccb806c328d48bc7b2 : SUCCESS in 53m 57s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/afbfd5f3571e465baf32d911f6260c6d : SUCCESS in 51m 44s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/caa8d990decf4c23b527257674aef0fe : FAILURE in 22m 51s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/e7d2c892c0b54812a003b51790f99f77 : FAILURE in 22m 59s","accounts_in_message":[],"_revision_number":5},{"id":"7ce5548ca1f1fd9711af115a60e9e3e01f4723c4","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-08 13:49:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c0f367dc780eac3015ac4dce8984482da5424917","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-08 13:53:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4865195352761ba01f687b0b4e63eb077bc481f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 14:50:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47249778975c43a7a9c39b4d661f2655\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a94f465493443ea11434ce693c23f1 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2664c1cac96e494fb5bb0aafdc3ab195 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a001a6b1601744aa89228f8947448b49 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52f1a8b792d84dd7a224f471d2388fa5 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0edde911c2ba4ae694964d7871c84a59 : SUCCESS in 7m 27s\n- aodh-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7e2b8ed834bc45e6b29341c350eb0808 : SUCCESS in 6m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f9eaccbc12be458bbb2460dceb53ee3c : SUCCESS in 46m 32s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/caf35db265d349d4a748c424c57c8ee3 : SUCCESS in 50m 50s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/655f390a54c943f483ee00f7e11c8f7e : SUCCESS in 42m 47s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6441f49772944ea8a11e7637249fc823 : SUCCESS in 51m 17s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/42fa6a316bd546a2ba6337920aaedbcd : FAILURE in 21m 06s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/8a02a064f39342d383171b87cf2395b1 : FAILURE in 23m 22s","accounts_in_message":[],"_revision_number":5},{"id":"040cc51df0e85ebf99454622d587476ac3c17b9a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-05-08 14:55:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2095ec52c7e946bd0d3d49f6da3d28e9eb195397","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-08 15:02:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7d997a9b67f70d47de6fc33b341c0d28ef184f93","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-08 15:47:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e4d9713a957e18ba47da02776c10cb87717aae83","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-09 01:11:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4145b2d9bc4497ae1d8f70f99dbe44c3b95c623e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 02:04:08.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a746d5be63d4420b3bf8fae84d39924\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd948d1d21de421da93ec431505887de : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ccc17c4f5bf4da6bc73bc35182d0d5a : SUCCESS in 7m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e19900ecfdda49b79eddefb4aab2bec7 : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f9dbf90905d41fb8e9e68832981201b : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e975704fe27c408185c6af68eb279c19 : SUCCESS in 10m 21s\n- aodh-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a8ffb8d36f88423ca1ce98c0e0377acb : SUCCESS in 6m 54s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/cfdd82f655cb49f2b0287491a3dd4c25 : SUCCESS in 45m 11s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/813afc238fb14c4aa4b1b71ee3467b62 : SUCCESS in 42m 13s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/93548eb07d6f4384bc15a0b862402460 : SUCCESS in 46m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4164e32d8bc043e6975a72acfd203259 : SUCCESS in 39m 23s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/54163ffd4a4448499f9d6aa5257f32f5 : FAILURE in 7m 26s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/55cc1f4554314dc2a9313571046e2ef2 : FAILURE in 9m 14s","accounts_in_message":[],"_revision_number":5},{"id":"9fe5df039217017783e4ee427f5b91a376272b3e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-05-09 09:55:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"69b5540e6cb12c00efe740e2d5dc4c4d112d8a98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 11:12:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de4cd4c15be849329042ff4331c05157\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea04af687baa4672becb77d81fbe7548 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63e738e59a3f43a58773a2738779f6bc : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1398ebe2a3d3482985b0b72c9aad7327 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8b9aa512b974635962ea048e87345ef : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29781b76a1fc44109172c02b71a37b6b : SUCCESS in 7m 19s\n- aodh-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/269291815a1e4206af430ba4b39704e8 : SUCCESS in 6m 49s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/db3db76669a8484aa8b1a11bc20bca4a : SUCCESS in 44m 31s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/58e69493df93425598e0f758fccb1736 : SUCCESS in 59m 09s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bbdca210dbdc4333879b92d2091a323b : SUCCESS in 53m 28s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a403a34a15934061aea3a88846c8d808 : SUCCESS in 51m 52s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/1d47b4c6dc0948cf84848033f1d381a7 : SUCCESS in 40m 47s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/aa30e33ebe6e46e5b7fa150e799e2f9f : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8fbdeae0b23172677319834cf092628f1e679de2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 11:13:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b19d8bd70be65ec2bfcf4397289355d742752d79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 12:18:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bed71c2e453431d88b860396253c4bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445f0e7ea8844948b77cc7668ef82218 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d81371a134a64bd6b8f1f03358023f75 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a406cc73f7124fd69bcbee6c9b33a5d7 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7a34699656442fb39af0957d86620e : SUCCESS in 6m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9b1f1b45ea55426b9a7ed993c2ac31d3 : SUCCESS in 46m 32s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1427352029dd41d195a85ceb0626837e : SUCCESS in 46m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/137e139a44ac46a6a23b02c76a729be9 : SUCCESS in 55m 47s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2cfff05e8a8e4ac992531aee34077577 : SUCCESS in 51m 00s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/4a6a6ee96bc14abc8653c8782736b308 : SUCCESS in 46m 08s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/f4a0e3ed61a44a2cb100a848fd2e8926 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":5},{"id":"8a1a195e01d86b5c1a033e87740f35ecd0b73ac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 12:18:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"eacb6098224ad1a66a1d269fe0a27a399db21645","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 12:19:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb6865d53c048078902cd2159903212\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9d0e1c55ee40e68ef7aed2efcda241 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6efa793ec2eaa02fa60894c994dbb761984f0702","revisions":{"f3990c6d2d48348f123fda6151e1b9f8551ddb98":{"kind":"REWORK","_number":1,"created":"2024-04-19 07:04:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/916382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/82/916382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/82/916382/1"}}},"commit":{"parents":[{"commit":"fa8b7dc862ca25d9719802fd2916002a98159d3d","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fa8b7dc862ca25d9719802fd2916002a98159d3d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 07:03:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 07:03:59.000000000","tz":540},"subject":"Fix db upgrade with SQLAlchemy 2.0","message":"Fix db upgrade with SQLAlchemy 2.0\n\nDB schema upgrade is stuck when SQLAlchemy 2.0 is used because all\noperations are not executed within a transaction.\n\nChange-Id: Ief7ea81e78344ea32fb2a295f11a5d5a873251a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f3990c6d2d48348f123fda6151e1b9f8551ddb98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f3990c6d2d48348f123fda6151e1b9f8551ddb98"}]},"branch":"refs/heads/master"},"50cc26447601bc4397e355850ae7b9a1a2fe73cc":{"kind":"REWORK","_number":2,"created":"2024-04-19 07:06:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/916382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/82/916382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/82/916382/2"}}},"commit":{"parents":[{"commit":"fa8b7dc862ca25d9719802fd2916002a98159d3d","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fa8b7dc862ca25d9719802fd2916002a98159d3d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 07:03:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 07:06:02.000000000","tz":540},"subject":"Fix db upgrade with SQLAlchemy 2.0","message":"Fix db upgrade with SQLAlchemy 2.0\n\nDB schema upgrade is stuck when SQLAlchemy 2.0 is used because all\noperations are not executed within a transaction.\n\nChange-Id: Ief7ea81e78344ea32fb2a295f11a5d5a873251a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/50cc26447601bc4397e355850ae7b9a1a2fe73cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/50cc26447601bc4397e355850ae7b9a1a2fe73cc"}]},"branch":"refs/heads/master"},"9fa033f38e0f69cdfe1d842e7a4427210a9b63a6":{"kind":"REWORK","_number":3,"created":"2024-04-22 14:20:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/916382/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/82/916382/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/82/916382/3"}}},"commit":{"parents":[{"commit":"fa8b7dc862ca25d9719802fd2916002a98159d3d","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fa8b7dc862ca25d9719802fd2916002a98159d3d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 07:03:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-22 14:12:40.000000000","tz":540},"subject":"Fix db upgrade with SQLAlchemy 2.0","message":"Fix db upgrade with SQLAlchemy 2.0\n\nDB schema upgrade is stuck when SQLAlchemy 2.0 is used because all\noperations are not executed within a transaction.\n\nChange-Id: Ief7ea81e78344ea32fb2a295f11a5d5a873251a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9fa033f38e0f69cdfe1d842e7a4427210a9b63a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9fa033f38e0f69cdfe1d842e7a4427210a9b63a6"}]},"branch":"refs/heads/master"},"0550aebf95a1f978b1cf89116132199b7560c431":{"kind":"REWORK","_number":4,"created":"2024-05-07 13:04:46.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/82/916382/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/82/916382/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/82/916382/4"}}},"commit":{"parents":[{"commit":"fa8b7dc862ca25d9719802fd2916002a98159d3d","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fa8b7dc862ca25d9719802fd2916002a98159d3d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 07:03:48.000000000","tz":540},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-05-07 13:04:42.000000000","tz":-240},"subject":"Fix db upgrade with SQLAlchemy 2.0","message":"Fix db upgrade with SQLAlchemy 2.0\n\nDB schema upgrade is stuck when SQLAlchemy 2.0 is used because all\noperations are not executed within a transaction.\n\nChange-Id: Ief7ea81e78344ea32fb2a295f11a5d5a873251a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0550aebf95a1f978b1cf89116132199b7560c431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0550aebf95a1f978b1cf89116132199b7560c431"}]},"branch":"refs/heads/master"},"6efa793ec2eaa02fa60894c994dbb761984f0702":{"kind":"REWORK","_number":5,"created":"2024-05-07 13:24:35.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/82/916382/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/82/916382/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/82/916382/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/82/916382/5"}}},"commit":{"parents":[{"commit":"fa8b7dc862ca25d9719802fd2916002a98159d3d","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fa8b7dc862ca25d9719802fd2916002a98159d3d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 07:03:48.000000000","tz":540},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-05-07 13:24:32.000000000","tz":-240},"subject":"Fix db upgrade with SQLAlchemy 2.0","message":"Fix db upgrade with SQLAlchemy 2.0\n\nDB schema upgrade is stuck when SQLAlchemy 2.0 is used because all\noperations are not executed within a transaction.\n\nChange-Id: Ief7ea81e78344ea32fb2a295f11a5d5a873251a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6efa793ec2eaa02fa60894c994dbb761984f0702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6efa793ec2eaa02fa60894c994dbb761984f0702"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
