)]}'
{"id":"openstack%2Faodh~916392","triplet_id":"openstack%2Faodh~master~I6931d274dbe5e3edcbaba7c621744c3a83acc364","project":"openstack/aodh","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-07 14:47:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6931d274dbe5e3edcbaba7c621744c3a83acc364","subject":"DNM: Debugging","status":"ABANDONED","created":"2024-04-19 08:40:40.000000000","updated":"2024-05-07 14:47:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1551a15427160a194dd8153a0c6fd6e07652eec5","_number":916392,"virtual_id_number":916392,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-23 07:33:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 11:51:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"077258a00ef42319b9070a689436b5ce89c6aebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 08:40:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3646cef9ee7d67e90e347c916e5cb79ed7d4cfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 09:32:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bfc34a3e25b8e3eaac1aa3f33f0d42bdfd2178ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 11:51:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ab4be3365ab4b6fa8be7484a8889fe1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06683e6fd82840f7b9b584191ee2bbd7 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fee5447efde8492e8ff1621cb8332c16 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49e4d4f007f24dbc9afebf16a82265af : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3266fb6e75ca49faa842704e249c7b7a : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4362f1be288f4bb2afecb945b50a35e6 : SUCCESS in 7m 17s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/217864bb9bef491eb00947cf459ba01a : TIMED_OUT in 2h 13m 16s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/12f8d47ed5bd48ca8074aeef2ea36faf : SUCCESS in 46m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/13f0551ef3804e63b07cb1dd8ab29c4a : TIMED_OUT in 2h 14m 14s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/141723bba54f4d0ab924742e672e0982 : TIMED_OUT in 2h 02m 31s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/d79e2893d022488f8f37907d49b4bf3b : SUCCESS in 58m 05s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/f65505f531674628807b8239381f2dcd : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c831fe72d49119551382aeb2112b14ca076db7b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-22 14:20:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c853e4cda8d40b4b5bf303f3623897303c929627","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-22 15:55:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a0b53e1ba0372a0c1fbff095273db256e4d24133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 18:14:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd3ae483b72544cb8e89573df635c459\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94a279c96a224ae7a30d0f09d8a13403 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8139fe3b503a4ba883240a2cedf044e1 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d212d5510cf4b9c9aad17edf7d4bb9f : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/216854e0250e4bc798b214b668b24eee : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6fea4d16444b63bad36ac4fc1ef333 : SUCCESS in 8m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/aca597ec07184cdd8491123530f442da : TIMED_OUT in 2h 12m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/dafbe94fcbd4410e98b740879a039850 : SUCCESS in 35m 46s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0e772edfe92f4746b4f3e3c1b6736980 : TIMED_OUT in 2h 12m 47s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d61c9785cd174872a89c4535ffe2b24d : TIMED_OUT in 2h 03m 01s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/16b98a572add4a6e8f2856e0a85f3617 : SUCCESS in 54m 02s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/d2d656c890e14198be5ec1493652ac31 : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":4},{"id":"1f21b5de0af09911342e95557232f74cf4cab940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 05:10:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"986d102f981258ca611a1fbece10cef4d7e6c07d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 07:33:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1775b152b80f470899ff117cbde1c683\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6298358097e94697b21d82506ca07aa2 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df1ffb2fb9ba4ac7b1d89039742714e5 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f251bd625d94bf7b618fafa0a0e8293 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95ed1d6a88ef4d3493abfd3a6f0db687 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b4e41e0c46421aa1a8176709f664f8 : SUCCESS in 6m 55s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8a071c89e3884667826173cd4db5b1be : TIMED_OUT in 2h 13m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/91632025a78845fd831bd745c201a98b : SUCCESS in 44m 48s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c21f12e73c364062bb861659e1e0dd93 : TIMED_OUT in 2h 14m 52s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0c8f2945cc9249d78c55f012914f47e7 : TIMED_OUT in 2h 04m 07s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/b6e764e1e8534ac08fed63b8f7f3ff17 : SUCCESS in 50m 35s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/3f98eb76219348848f687e1a12bfe713 : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":5},{"id":"1551a15427160a194dd8153a0c6fd6e07652eec5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 14:47:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a51d91af4684a73e97341db1b81f4aa0e19eb363","revisions":{"6f6e2528faf56aa2f6f2958c14f86c95c62ec69b":{"kind":"REWORK","_number":1,"created":"2024-04-19 08:40:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/916392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/92/916392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/92/916392/1"}}},"commit":{"parents":[{"commit":"13c56221709dbeef71cc3e454c033a292b0d3542","subject":"Remove SQLAlchemy tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/13c56221709dbeef71cc3e454c033a292b0d3542"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 08:40:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 08:40:23.000000000","tz":540},"subject":"DNM: debugging ...","message":"DNM: debugging ...\n\nChange-Id: I6931d274dbe5e3edcbaba7c621744c3a83acc364\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f6e2528faf56aa2f6f2958c14f86c95c62ec69b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f6e2528faf56aa2f6f2958c14f86c95c62ec69b"}]},"branch":"refs/heads/master"},"b008cd14d95a793fb0e94b65a7f6581a154fb639":{"kind":"REWORK","_number":2,"created":"2024-04-19 09:32:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/916392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/92/916392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/92/916392/2"}}},"commit":{"parents":[{"commit":"13c56221709dbeef71cc3e454c033a292b0d3542","subject":"Remove SQLAlchemy tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/13c56221709dbeef71cc3e454c033a292b0d3542"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 09:32:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 09:32:17.000000000","tz":540},"subject":"DNM: Debugging","message":"DNM: Debugging\n\nChange-Id: I6931d274dbe5e3edcbaba7c621744c3a83acc364\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b008cd14d95a793fb0e94b65a7f6581a154fb639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b008cd14d95a793fb0e94b65a7f6581a154fb639"}]},"branch":"refs/heads/master"},"6c5dca0891c918d250b34e95c8e5ec584c1def8d":{"kind":"REWORK","_number":3,"created":"2024-04-22 14:20:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/916392/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/92/916392/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/92/916392/3"}}},"commit":{"parents":[{"commit":"93dd63bb8984d99a59fda0d34e7955f82305b0f9","subject":"Remove SQLAlchemy tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93dd63bb8984d99a59fda0d34e7955f82305b0f9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 09:32:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-22 14:19:42.000000000","tz":540},"subject":"DNM: Debugging","message":"DNM: Debugging\n\nChange-Id: I6931d274dbe5e3edcbaba7c621744c3a83acc364\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6c5dca0891c918d250b34e95c8e5ec584c1def8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6c5dca0891c918d250b34e95c8e5ec584c1def8d"}]},"branch":"refs/heads/master"},"5c85e8e19f25b556b8dadd018f5844f50168a921":{"kind":"REWORK","_number":4,"created":"2024-04-22 15:55:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/916392/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/92/916392/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/92/916392/4"}}},"commit":{"parents":[{"commit":"d8d1bf432d99a5601739c26a81255a35e3d27371","subject":"Remove SQLAlchemy tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d8d1bf432d99a5601739c26a81255a35e3d27371"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 09:32:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-22 15:55:36.000000000","tz":540},"subject":"DNM: Debugging","message":"DNM: Debugging\n\nChange-Id: I6931d274dbe5e3edcbaba7c621744c3a83acc364\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5c85e8e19f25b556b8dadd018f5844f50168a921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5c85e8e19f25b556b8dadd018f5844f50168a921"}]},"branch":"refs/heads/master"},"a51d91af4684a73e97341db1b81f4aa0e19eb363":{"kind":"REWORK","_number":5,"created":"2024-04-23 05:10:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/916392/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/92/916392/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/92/916392/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/92/916392/5"}}},"commit":{"parents":[{"commit":"d8d1bf432d99a5601739c26a81255a35e3d27371","subject":"Remove SQLAlchemy tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d8d1bf432d99a5601739c26a81255a35e3d27371"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 09:32:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-23 05:10:28.000000000","tz":540},"subject":"DNM: Debugging","message":"DNM: Debugging\n\nChange-Id: I6931d274dbe5e3edcbaba7c621744c3a83acc364\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a51d91af4684a73e97341db1b81f4aa0e19eb363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a51d91af4684a73e97341db1b81f4aa0e19eb363"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
